)]}'
{"id":"openstack%2Fkeystone~806458","triplet_id":"openstack%2Fkeystone~master~Ia72302c2667d88aeecdb4ccda5d0899fa0639fef","project":"openstack/keystone","branch":"master","hashtags":[],"change_id":"Ia72302c2667d88aeecdb4ccda5d0899fa0639fef","subject":"Add some missing parameters in docs of oauth","status":"ABANDONED","created":"2021-08-28 03:45:56.000000000","updated":"2021-08-28 04:04:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3cbb78478e934ed6ce07dbc3e326c35b01281bfa","_number":806458,"virtual_id_number":806458,"owner":{"_account_id":33899,"name":"test","email":"531648642@qq.com","username":"bingdian"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-08-28 03:59:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-28 03:59:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2f61e394fc43219fa222ad0e3a8da0c0dbd8fa54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33899,"name":"test","email":"531648642@qq.com","username":"bingdian"},"date":"2021-08-28 03:45:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b260d2ab84db3030be20c36080dc43ccc2395bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-28 03:59:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cefb339da716409a9f19d1b13c0fd4b6 : FAILURE in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5309e6850074208b9ce5894757851c0 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9db9630bff845dc87343e521bda2b0d : SUCCESS in 12m 18s","accounts_in_message":[],"_revision_number":1},{"id":"3cbb78478e934ed6ce07dbc3e326c35b01281bfa","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33899,"name":"test","email":"531648642@qq.com","username":"bingdian"},"date":"2021-08-28 04:04:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6259ddd16f3868dfeb0da5b1034d719f031082a7","revisions":{"6259ddd16f3868dfeb0da5b1034d719f031082a7":{"kind":"REWORK","_number":1,"created":"2021-08-28 03:45:56.000000000","uploader":{"_account_id":33899,"name":"test","email":"531648642@qq.com","username":"bingdian"},"ref":"refs/changes/58/806458/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/58/806458/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/58/806458/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/58/806458/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/58/806458/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/58/806458/1"}}},"commit":{"parents":[{"commit":"8573e7dbf0911b99456e9dc77a1b27aeca813a6e","subject":"Add some missing parameters in docs of oauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8573e7dbf0911b99456e9dc77a1b27aeca813a6e"}]}],"author":{"name":"budebao","email":"531648642@qq.com","date":"2021-08-28 03:45:46.000000000","tz":480},"committer":{"name":"budebao","email":"531648642@qq.com","date":"2021-08-28 03:45:46.000000000","tz":480},"subject":"Add some missing parameters in docs of oauth","message":"Add some missing parameters in docs of oauth\n\nChange-Id: Ia72302c2667d88aeecdb4ccda5d0899fa0639fef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6259ddd16f3868dfeb0da5b1034d719f031082a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6259ddd16f3868dfeb0da5b1034d719f031082a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
