)]}'
{"id":"openstack%2Fkeystone~812990","triplet_id":"openstack%2Fkeystone~stable%2Fwallaby~Idfd9adbbe800bbc21814d94002a2b61524cce28a","project":"openstack/keystone","branch":"stable/wallaby","hashtags":[],"change_id":"Idfd9adbbe800bbc21814d94002a2b61524cce28a","subject":"Fix typo in identity provider policies","status":"MERGED","created":"2021-10-07 13:24:56.000000000","updated":"2021-10-07 19:52:49.000000000","submitted":"2021-10-07 19:50:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812990","meta_rev_id":"edece369abf8789c6d8e7d9434ef734ac9bc96ae","_number":812990,"virtual_id_number":812990,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-07 19:50:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"recommended":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2021-10-07 13:52:02.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2021-10-07 18:10:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2021-10-07 18:10:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-07 13:52:02.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2021-10-07 14:45:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-07 15:00:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0b25913efe7d93424fc3023b947545a45607a661","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-10-07 13:24:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7eacc4896bfb6cd05c23a224c4f76899ae58514","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2021-10-07 13:52:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"35c58c2d2644b43b6dba7407aad93ca6ee0c9a35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 14:45:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/400d6010362048f59eb87e3d6e4d3747 : TIMED_OUT in 1h 01m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8fb936301dce4b1faf7a519d557d5a78 : TIMED_OUT in 1h 01m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0acbf33e49204a505769ca0d4dded8094594b24b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 15:00:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dcba933d9b4a4a3e9c00163cc7c79392 : TIMED_OUT in 53m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9721d8c035824e0aaad5d67d46ec0a8f : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0539c91a73254816b834fcfdddde59e4 : SUCCESS in 13m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5ffe6ef6f28462f8b8f1db21f0e8570 : SUCCESS in 15m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bf2bb79010048bcba4aa449cc06a3b7 : SUCCESS in 13m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/44fa9812abc34fea93d3b2849077c761 : SUCCESS in 1h 04m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1f52759e1f4946768fe371be5a290a6a : SUCCESS in 1h 29m 53s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/922f7c6543144e99b13e5fbdc4309138 : SUCCESS in 41m 41s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31d78bf1c5884140ae39cb2679f9e86d : SUCCESS in 35m 43s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/4bc1ea3351ac400d8df1ead02cc0d190 : SUCCESS in 39m 10s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/e2ac08af7c1a4efab08f48a98e7bcc38 : SUCCESS in 22m 37s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/e7a4c7d527e540a3b55c250bb0a2ea80 : FAILURE in 19m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfd99d9226a44ae68f27030a8912a9ac : SUCCESS in 1h 02m 19s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/a188c6feab984effae65d69d28b49640 : FAILURE in 59m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0c3c86bc90436b52335a6c573bac69b2049c7b56","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-10-07 16:43:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5cdc15157b05ff84fc7e055e0673bab514f180bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 18:09:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca0e119144074e769958ea00cb7aa2bb : SUCCESS in 17m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3e961d821e74812803b1b5191bb651e : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e44d5557f47e4b339487ee83dacc91b5 : SUCCESS in 19m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4195d59023bd45eb87dd8a28abcea66c : SUCCESS in 18m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/781c8a7309d34c9b99fe535d3177cd39 : SUCCESS in 14m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/7d7ac2c3199f479ca2aa6b8aba581552 : SUCCESS in 1h 01m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff1ea66654554e9c8c4a489030b29e1e : SUCCESS in 1h 22m 34s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/7fe15dde55ac4f87b7f21758be467776 : SUCCESS in 46m 25s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56f8d94aaa3b4010ad5832f2e8aad256 : SUCCESS in 30m 59s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/75319f7e3f1c4690b033d47a194a0c10 : SUCCESS in 37m 26s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/be308c32cda44c738cb035367669bdc6 : SUCCESS in 14m 26s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/d15478a2a9ed43baaf9bcb11126e88d1 : FAILURE in 19m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2920710f1aa144f69a42692c6273e11c : SUCCESS in 56m 37s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/37fce5b0ce3c45718acd53d81b4b0853 : SUCCESS in 42m 18s","accounts_in_message":[],"_revision_number":1},{"id":"4dc7eeebe89f1fa8ab2a5dcff3dca58810921b69","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-10-07 18:10:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e9067d9c182e409f1a32511006eb8f0e045790d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 18:11:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"712db3e6f440a8438470a3687d5456637b0fcf68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 19:18:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c9dfc43c5cb47e8a675dba71d56d7d9 : TIMED_OUT in 1h 01m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff4e251cf02246a8955bbe1e33923004 : TIMED_OUT in 1h 01m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"23481a3887370264cdb946fd8dd5561b564d3d6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 19:50:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/912ef70e05b84263a1b3301cebfedc12 : SUCCESS in 8m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15a57aabb5ed4433806ad8260f192a16 : SUCCESS in 20m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72b4380b2e254ab3910d6bfa7839aaec : SUCCESS in 16m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cff08543ef6241acae98f34cb0320609 : SUCCESS in 15m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/d603eac52f3a49bf95b558e536bedd9a : SUCCESS in 59m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ed63d89fbdb499989a90f3b2ec3d532 : SUCCESS in 1h 17m 06s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/cc6ff5c38128477c97d0f468f72bcff3 : SUCCESS in 31m 18s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/2e559d8056934802aca023e62a169c26 : SUCCESS in 31m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e0e0127b627415b8df5419cf38f544a : SUCCESS in 46m 11s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/a4ae090073c34a83ab7f6d0a97c579bc : SUCCESS in 40m 13s","accounts_in_message":[],"_revision_number":1},{"id":"713fdeb131302124c82ed47a02d25b8e30a39932","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 19:50:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"edece369abf8789c6d8e7d9434ef734ac9bc96ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 19:52:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ece3efa8379148a6aada768f73b2b6a0 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bdd8f82f60d2e46e5f6951c4407366b89591cde5","revisions":{"bdd8f82f60d2e46e5f6951c4407366b89591cde5":{"kind":"REWORK","_number":1,"created":"2021-10-07 13:24:56.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/90/812990/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/90/812990/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/90/812990/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/90/812990/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/90/812990/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/90/812990/1"}}},"commit":{"parents":[{"commit":"f510c806de3e20cdedd55291cd58dafa59398bec","subject":"Hide AccountLocked exception from end users","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f510c806de3e20cdedd55291cd58dafa59398bec"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-26 16:29:51.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-10-07 13:23:52.000000000","tz":0},"subject":"Fix typo in identity provider policies","message":"Fix typo in identity provider policies\n\nThis cause the sample generated policy file to alias the old name with\nthe new policy name, which isn\u0027t needed since we\u0027re not renaming these\npolicies at all and it was likely a typo.\n\nConflicts:\n      keystone/common/policies/identity_provider.py\n      In later releases the deprecation parameters were moved up to the\n      deprecated options and not in the DocumentedRule defaults.\n\nChange-Id: Idfd9adbbe800bbc21814d94002a2b61524cce28a\n(cherry picked from commit c10d5c88ef40e63d4dfefb792d6c3d68acd72dd9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bdd8f82f60d2e46e5f6951c4407366b89591cde5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bdd8f82f60d2e46e5f6951c4407366b89591cde5"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
