)]}'
{"id":"openstack%2Fkeystone~812994","triplet_id":"openstack%2Fkeystone~stable%2Fstein~Idfd9adbbe800bbc21814d94002a2b61524cce28a","project":"openstack/keystone","branch":"stable/stein","hashtags":[],"change_id":"Idfd9adbbe800bbc21814d94002a2b61524cce28a","subject":"Fix typo in identity provider policies","status":"MERGED","created":"2021-10-07 13:28:13.000000000","updated":"2021-10-12 05:02:53.000000000","submitted":"2021-10-12 05:01:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812994","meta_rev_id":"f9885cdb2874b67ce960924ea592805198659ac9","_number":812994,"virtual_id_number":812994,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-12 05:01:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"recommended":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2021-10-07 13:52:50.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2021-10-12 03:24:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2021-10-12 03:24:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-07 13:52:50.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2021-10-07 14:29:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"97ab511ac1c3ca82092b2a3d7b7c237ab2d5ad40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-10-07 13:28:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cbe209c3ac410b6f42ab447ad968cce63d54ad83","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2021-10-07 13:52:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bfb122d22b55bec17d3f32710a687a278487261c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 14:29:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4773bf8b1cc046c6be3f54e0d547e2f4 : SUCCESS in 42m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3efdbe19369e459ba37ecafe972dcac0 : SUCCESS in 38m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbd610402a00493dbc2039a21320b327 : SUCCESS in 7m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/628137ae19bb49d0a6f417bc9c2de24d : SUCCESS in 21m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01d7ea1c96a24b5eb4ffe95eafce3eff : SUCCESS in 24m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a6fb39620058400b84fa902f6beb480d : SUCCESS in 36m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aadee413a2ce425595712f504a9895b8 : SUCCESS in 10m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5b1c58ff703c41faacf3f62dbdcf13b2 : FAILURE in 31m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bf4f2a8a07b74a48a8af6be108d678a8 : SUCCESS in 55m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb4b5914a5654c7d9a131a739eb32235 : FAILURE in 36m 34s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/63dc7acad7b744f3adbf2090ddff638b : FAILURE in 22m 35s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/cf7743d1888542088fa8a2b68f2c934f : FAILURE in 34m 48s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/7a221b8e9a1f428d865cb24e25012ae4 : FAILURE in 10m 51s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/62ce1c2a96ff4ea0b0dde006079fe1ae : FAILURE in 12m 05s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/d74921a4bb36401aadce9f140938718a : FAILURE in 11m 28s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/a0046b9782a3426f8d57b5a3eeaf9e27 : SUCCESS in 20m 50s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/39ac4bb7000b436e812dae3b5ba134a8 : FAILURE in 26m 07s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a0288f7a07974dcab36a03e29722d971 : SUCCESS in 52m 04s","accounts_in_message":[],"_revision_number":1},{"id":"4f544b224f8523bd0ae712c3856ec591c902a9d2","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-10-11 19:59:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c7eb3ca1644d076b6aecf300afd294ec13a93932","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 21:29:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b9a3b3c4e564b0aba372a342a38adb0 : SUCCESS in 32m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/206160ba29e14b77b4f6bbcf45f02995 : SUCCESS in 25m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d427d6f8d490458b87f84cd9c44d818d : SUCCESS in 7m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cb9b5e09d2804502a488b6d5d63095c0 : SUCCESS in 40m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcd58bd2ec72478ba21d4eec5b7bbe6b : SUCCESS in 29m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b20f151e1554d4db821941b0a708b8d : SUCCESS in 30m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c690079a38b24d54856263318f1a4b4e : SUCCESS in 8m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6c5b8f17327e41c4bb8f3e395f1b8a22 : SUCCESS in 1h 21m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/aa0e83bce1594a988caa205ea17bc2f8 : SUCCESS in 54m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f28c549a6d7f4697813e85be71372f82 : SUCCESS in 1h 15m 13s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/c795078ae3dd43e58cf13c30b37cb57e : SUCCESS in 37m 51s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/36b895a67f2f402a9aaae51b2851f002 : SUCCESS in 31m 56s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/9d83af116d2c4c5a953133f81907031a : FAILURE in 13m 03s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/713dcec8131b42dba87e1d3a7407e999 : FAILURE in 13m 02s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/1effe9c52eb24b2abc51ab67b0b9d1cd : FAILURE in 15m 12s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/490236861e4341bab7725efb3522a267 : SUCCESS in 18m 33s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/0900b1e416e849cfb820fc0fed7ba551 : SUCCESS in 34m 44s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/635dff8c46b14dc8bc4c1d3327c95f2a : SUCCESS in 48m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f95d03e89490d5fef8b91860a4207f7163c0419c","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-10-12 03:24:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2ef42dc27fe09074326d3a1bbf4443f398f872dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 03:25:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b421dd200ff84913a7a55a7748f55d00566b4463","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 05:01:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5cfe2bd1ea934097aef2aad05197b83f : SUCCESS in 31m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/072594736b474524ac0e847d20dd7eee : SUCCESS in 7m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/875533b357dd45d9b5573e0a3ee82b37 : SUCCESS in 23m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/770a8c28665644a8b0f414bd89207d97 : SUCCESS in 32m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13ea07af2d4f464cb0c8fcb4d5812fdf : SUCCESS in 33m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37ce6a307f0e4f538885b9633cce65e5 : SUCCESS in 10m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0b6e855c98e64f5e96eb880ee0cc8cd7 : SUCCESS in 1h 30m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4d61068bc205444794fef04687c1da3f : SUCCESS in 1h 16m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2ba9f01cbc14a08bed88599633a2823 : SUCCESS in 1h 23m 13s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/f2577c4d1acd4435b853c6c127054bfa : SUCCESS in 39m 58s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/5427b4e0bcbe45f9b7fc566b6dbc84fa : SUCCESS in 41m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e2eef9d11f91468f821b7c58761a7a29 : SUCCESS in 55m 30s","accounts_in_message":[],"_revision_number":1},{"id":"1a600c97c5c90a64a89315c04f39a5220789c94b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 05:01:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f9885cdb2874b67ce960924ea592805198659ac9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 05:02:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33116d358d9547cc86749806d04060dd : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f0bf1058fa0df357b68aa54353e44d3708c24ec0","revisions":{"f0bf1058fa0df357b68aa54353e44d3708c24ec0":{"kind":"REWORK","_number":1,"created":"2021-10-07 13:28:13.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/94/812994/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/94/812994/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/94/812994/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/94/812994/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/94/812994/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/94/812994/1"}}},"commit":{"parents":[{"commit":"f36034c8a6530b16f6b6eb88ee97f540c974ba00","subject":"Retry update_user when sqlalchemy raises StaleDataErrors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f36034c8a6530b16f6b6eb88ee97f540c974ba00"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-26 16:29:51.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-10-07 13:27:35.000000000","tz":0},"subject":"Fix typo in identity provider policies","message":"Fix typo in identity provider policies\n\nThis cause the sample generated policy file to alias the old name with\nthe new policy name, which isn\u0027t needed since we\u0027re not renaming these\npolicies at all and it was likely a typo.\n\nConflicts:\n      keystone/common/policies/identity_provider.py\n      In later releases the deprecation parameters were moved up to the\n      deprecated options and not in the DocumentedRule defaults.\n\nChange-Id: Idfd9adbbe800bbc21814d94002a2b61524cce28a\n(cherry picked from commit c10d5c88ef40e63d4dfefb792d6c3d68acd72dd9)\n(cherry picked from commit bdd8f82f60d2e46e5f6951c4407366b89591cde5)\n(cherry picked from commit f742fadef1718e070ff4e151d400a96dc6acf74e)\n(cherry picked from commit 3fd71f5a933052b0a3fd4dadb3a754ac544348f3)\n(cherry picked from commit db13e36e7f95c770f0a52c57e89fc54c72faf2eb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f0bf1058fa0df357b68aa54353e44d3708c24ec0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f0bf1058fa0df357b68aa54353e44d3708c24ec0"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
