)]}'
{"id":"openstack%2Fkeystone~812998","triplet_id":"openstack%2Fkeystone~stable%2Fussuri~I54021986d17c57d7733d53caa4032c2767eaf25e","project":"openstack/keystone","branch":"stable/ussuri","hashtags":[],"change_id":"I54021986d17c57d7733d53caa4032c2767eaf25e","subject":"Fix typos in ec2 credential policies","status":"MERGED","created":"2021-10-07 13:34:28.000000000","updated":"2021-10-11 16:59:00.000000000","submitted":"2021-10-11 16:56:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812998","meta_rev_id":"85e7af85404fe6a61a1f4af2a1865f28c40fd3cc","_number":812998,"virtual_id_number":812998,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-11 16:56:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"recommended":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2021-10-07 13:50:48.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2021-10-07 13:44:35.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2021-10-11 13:17:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2021-10-11 13:17:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-07 13:44:35.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2021-10-07 13:50:48.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2021-10-07 14:58:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"508eef52b090c0899408538e4cf2dbd8860267a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-10-07 13:34:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6dc1e5903f0579a05cc5943bbce7518b46b300e","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2021-10-07 13:44:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c2a0b91ece29082934d1e6bc6b54a066aabfd195","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2021-10-07 13:50:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e20fe43a19a8204baa1ba3099dd84728b94603db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 14:58:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19ae04ee7b3b437687f1eefb74a16c1a : SUCCESS in 23m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a710110e89e34b75953ff6ff0d8c68a1 : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44f58c2a989649589592030764106b8c : SUCCESS in 20m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f4e13684bf2248b2801881fc07ba5c3e : SUCCESS in 18m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93531c5f71414e6791d3036ad68a1c47 : SUCCESS in 13m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/0b873793c6804e609e76a32b4962715d : SUCCESS in 1h 11m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/54fd024bcf0c433ea0191b7721db4696 : SUCCESS in 1h 21m 22s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/86f25b42b95e471eaa87d2787eb82a81 : SUCCESS in 48m 49s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/86087489be4c44249fc3233a92577783 : SUCCESS in 35m 43s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/e20ad28e63184127baf75f37c18fea88 : SUCCESS in 33m 14s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/87cec32c218b466b973bc88d6e50a01c : FAILURE in 12m 41s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/db63bf8fcca94c87bdcbf8f789176c78 : SUCCESS in 35m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78bdc11be9ee4bea89a6721649f481ff : SUCCESS in 47m 16s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/72da554351b44ae296df73aa392419e1 : SUCCESS in 35m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c6ed285494d1254096fea07cc66a6fce3a09fe37","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-10-11 13:17:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"01386458d9347ea7c747d07a7c9d6647d10c351d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 13:18:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"62c89241bd765f1052c32d6276a3119f9d3225a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 16:56:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd3fa110c7304a14b2a104f768076a35 : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/232c708f05a947bfbbe13af713bc9ff9 : SUCCESS in 13m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78468899d0794957a9481b578862d28e : SUCCESS in 15m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd0984aa61414260b0c50adb8f0ea17b : SUCCESS in 13m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/c40ca4b7e698401f8c417e93b716d73a : SUCCESS in 1h 10m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7cba5e5561b9402e84691f05ba9504a4 : SUCCESS in 1h 24m 27s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/ce7dbdbf4ece44e89497092603afae49 : SUCCESS in 30m 04s\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/5a88f6c566ba4de19b733024e1f359c7 : SUCCESS in 35m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87b1735f8dd448e8ae43010927fde552 : SUCCESS in 55m 48s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/44db5300269b4d1db45281fd860febd5 : SUCCESS in 34m 53s","accounts_in_message":[],"_revision_number":1},{"id":"6692dcc4d6dc142d260de3432dd75ca6e6e62d88","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 16:56:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"85e7af85404fe6a61a1f4af2a1865f28c40fd3cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 16:59:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c8ad200e1b048469c598f0efcea8f7e : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b38928d0a26e8ab4c1496e2eaea7598cf72e5b37","revisions":{"b38928d0a26e8ab4c1496e2eaea7598cf72e5b37":{"kind":"REWORK","_number":1,"created":"2021-10-07 13:34:28.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/98/812998/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/98/812998/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/98/812998/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/98/812998/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/98/812998/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/98/812998/1"}}},"commit":{"parents":[{"commit":"2cf2912fd4d23bd8515313eafc77c6df92c68b81","subject":"Merge \"Hide AccountLocked exception from end users\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2cf2912fd4d23bd8515313eafc77c6df92c68b81"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-26 14:50:35.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-10-07 13:34:10.000000000","tz":0},"subject":"Fix typos in ec2 credential policies","message":"Fix typos in ec2 credential policies\n\nThere was a trailing s in two of these policies and it caused the policy\nnames to mismatch, which causes confusion with the rendered policy files\nand potentially causes uses with deprecation logic.\n\nChange-Id: I54021986d17c57d7733d53caa4032c2767eaf25e\n(cherry picked from commit 82da8824df0f56ef4e137805bf32d647cef1ea59)\n(cherry picked from commit 6dff22b5baa1a19842aca435782fe1f9789f72cc)\n(cherry picked from commit a57ae85c9699e7a560b7ffe9786ed0a6453c1e86)\n(cherry picked from commit 65c99d6efb6a83fcc16ddf957e59297763e85b6c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b38928d0a26e8ab4c1496e2eaea7598cf72e5b37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b38928d0a26e8ab4c1496e2eaea7598cf72e5b37"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
