)]}'
{"id":"openstack%2Fkeystone~812999","triplet_id":"openstack%2Fkeystone~stable%2Ftrain~I54021986d17c57d7733d53caa4032c2767eaf25e","project":"openstack/keystone","branch":"stable/train","hashtags":[],"change_id":"I54021986d17c57d7733d53caa4032c2767eaf25e","subject":"Fix typos in ec2 credential policies","status":"MERGED","created":"2021-10-07 13:34:54.000000000","updated":"2021-10-11 22:01:35.000000000","submitted":"2021-10-11 21:59:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812999","meta_rev_id":"7e16ba497700cc4e30a579a35da6cd6b0566b11f","_number":812999,"virtual_id_number":812999,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-11 21:59:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"recommended":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2021-10-07 13:50:54.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2021-10-07 13:44:19.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2021-10-11 19:58:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2021-10-11 19:58:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-07 13:44:19.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2021-10-07 13:50:54.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2021-10-07 15:03:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d748d6260337e44563cbdecb26de83efd7537ee0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-10-07 13:34:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1949c812f66ed60fbbb336e1096d5736e9c39daf","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2021-10-07 13:44:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e44da36085738cf9cba2e2f18b0528cc450a3bf0","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2021-10-07 13:50:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1e8576ef8fbf9161d8f1d1577594d41bdcfad7d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 15:03:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71b4327efa484df9824fa506cea0ca7b : SUCCESS in 15m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d75a116585d4b32a230803357a57124 : SUCCESS in 8m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f7ae2e10b02e4a459515ab0cbff0f97d : SUCCESS in 17m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b092df4216be4528b88f7ed5248bddcd : SUCCESS in 20m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4870e86efcea4201a2eaac4b89391ec6 : SUCCESS in 14m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f0c9d5578074d76be8f09a4c8c26b3f : SUCCESS in 15m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/76fe974ac1e44e388a7d93fcfa8ed029 : SUCCESS in 1h 25m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/56634a577d4647ef867d2e9a7bfa8c42 : FAILURE in 39m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/15012a5d2501415db4c5585a46f68d8f : FAILURE in 31m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4f939a9dc7a48af90c92d39c68ec8cd : SUCCESS in 1h 23m 26s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/17671a0ff28848c683638b48ad2d0a39 : SUCCESS in 48m 26s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/7ba421b34a3c4f4fa186875e5d44b466 : SUCCESS in 42m 06s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/74561d69fd19409792bbe94a6e5bd837 : FAILURE in 8m 59s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/9eb4c42a423f4426a031642b2dbe320c : FAILURE in 9m 44s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/370681afcede498ba29e7a227e427fdd : FAILURE in 8m 59s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/113dea0dcdae43eaae80f701df782d7f : SUCCESS in 29m 37s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/0b775bc45b454b7caaa1870a95496a03 : SUCCESS in 46m 07s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6777eadbaf764efd8dd27e2793f45316 : FAILURE in 30m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5510bd8310e434cb106e7f41004fcc3 : SUCCESS in 53m 01s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/e352198186784eaeaf9aa83f8a76d593 : SUCCESS in 41m 22s","accounts_in_message":[],"_revision_number":1},{"id":"5b2dc06ee62df557e4210e2cf09bc1ffa98e1fe6","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-10-11 17:34:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9be9f05684fb4b1cba0a305ee39ebb4f357aac53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 19:28:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c76e8836c2bd41728b2441f2962d78d0 : SUCCESS in 21m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab673143c5cd45d0b059e8b942a080ce : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/06a8f6c0fca34d7898b9664a91658e04 : SUCCESS in 14m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2493e5307cd44f12840b80260e85afc2 : SUCCESS in 16m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/546e07b6b42e48bea663053f357c74c2 : SUCCESS in 17m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3e4d82a6b8e47fd8df32c3ee7eb499a : SUCCESS in 15m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a6a392a75fdc45c397bab6fac06b5c77 : SUCCESS in 1h 43m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8d50abc1d1714cd1b148132fa05b44a9 : SUCCESS in 50m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/fd4e53f00c0b4b8c9b435dafcf1fd0cb : SUCCESS in 1h 04m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1aad3d48340a4772a923b627be252833 : SUCCESS in 1h 47m 36s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/56ad30e8687346098c77fcac353e5c8b : SUCCESS in 25m 30s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/bd04f156f00e40a89ba70956c060e02c : SUCCESS in 31m 59s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/b5ce0db24b2040bf98af44c6508906c8 : FAILURE in 13m 22s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/1a7c4656eeec4b018893ed8295140598 : FAILURE in 8m 46s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/ce3ba3cf06694065b298ca69aa7ca678 : FAILURE in 14m 16s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/5f48b05f63294226adb16606eecadf3e : SUCCESS in 13m 18s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/f31109f9594a44c6ae58d1975b7e3598 : SUCCESS in 36m 49s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fcda42c140cd4bfa8bb15c05b919f362 : SUCCESS in 1h 17m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fcf8684580f8480ab78278f87495b801 : SUCCESS in 53m 27s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/63d0866d0bcb45ac910fc18b54f267f4 : SUCCESS in 53m 27s","accounts_in_message":[],"_revision_number":1},{"id":"19afe5f2390283826818922d7207c341d4d5302c","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-10-11 19:58:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8a0d996ef8bfb0a83f224eaa31e84fdbaf1f887a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 20:00:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9799ffa14f81f470734dea6d15484bf13fd1b1b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 21:59:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9e7b54950d6426686d3f79920d4ecb8 : SUCCESS in 7m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/90b761a2eac04166a1def7ca69b9cfbc : SUCCESS in 13m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84eadf1bed4d44b1b7a49ac354725fd3 : SUCCESS in 15m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/40faf95feaf74f2ca91b61106d6881d9 : SUCCESS in 14m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cac12eeeeac54628bec3eac7491caf33 : SUCCESS in 13m 29s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e33fef8536e24fdba8516e73304ba177 : SUCCESS in 1h 57m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9c474a9a8d4947388c8b800e8ba6b9b0 : SUCCESS in 47m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/071ec20d44cf422a913105e34d3b74bd : SUCCESS in 57m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5199432f8633442dbd639a5b8dfd727b : SUCCESS in 1h 21m 02s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/7784aac9a206465d8c6b878ec3eee26c : SUCCESS in 23m 23s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/16a9718877e64961b94c522d8a5d8670 : SUCCESS in 31m 36s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3a121805000b438c979bccb482da093f : SUCCESS in 55m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9467bcfc78984c75b2803e86d2033efe : SUCCESS in 48m 50s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/f4995d8824974b0d854bb3565b30f611 : SUCCESS in 33m 10s","accounts_in_message":[],"_revision_number":1},{"id":"0de4baf3703fae7636f1080269065c59c7149430","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 21:59:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7e16ba497700cc4e30a579a35da6cd6b0566b11f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 22:01:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/677d5ba48c9f4828814d3e0d9e72f807 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1558adc73d363949aa2b696c0fc7b8a249228072","revisions":{"1558adc73d363949aa2b696c0fc7b8a249228072":{"kind":"REWORK","_number":1,"created":"2021-10-07 13:34:54.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/99/812999/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/99/812999/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/99/812999/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/99/812999/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/99/812999/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/99/812999/1"}}},"commit":{"parents":[{"commit":"acef9c60722edf78bcb85328ca5ab23331ab9273","subject":"Remove group deletion for non-sql driver when removing domains.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/acef9c60722edf78bcb85328ca5ab23331ab9273"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-08-26 14:50:35.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-10-07 13:34:52.000000000","tz":0},"subject":"Fix typos in ec2 credential policies","message":"Fix typos in ec2 credential policies\n\nThere was a trailing s in two of these policies and it caused the policy\nnames to mismatch, which causes confusion with the rendered policy files\nand potentially causes uses with deprecation logic.\n\nChange-Id: I54021986d17c57d7733d53caa4032c2767eaf25e\n(cherry picked from commit 82da8824df0f56ef4e137805bf32d647cef1ea59)\n(cherry picked from commit 6dff22b5baa1a19842aca435782fe1f9789f72cc)\n(cherry picked from commit a57ae85c9699e7a560b7ffe9786ed0a6453c1e86)\n(cherry picked from commit 65c99d6efb6a83fcc16ddf957e59297763e85b6c)\n(cherry picked from commit b38928d0a26e8ab4c1496e2eaea7598cf72e5b37)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1558adc73d363949aa2b696c0fc7b8a249228072"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1558adc73d363949aa2b696c0fc7b8a249228072"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
