)]}'
{"id":"openstack%2Fkeystone~815168","triplet_id":"openstack%2Fkeystone~master~I3be107406d57545db4e49bb705b82c3666b06436","project":"openstack/keystone","branch":"master","attention_set":{"33826":{"account":{"_account_id":33826,"name":"Seena Fallah","email":"seenafallah@gmail.com","username":"clwluvw"},"last_update":"2021-12-08 16:19:57.000000000","reason":"Lance Bragstad replied on the change"}},"removed_from_attention_set":{"5046":{"account":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"last_update":"2021-12-08 16:19:57.000000000","reason":"Lance Bragstad replied on the change"}},"hashtags":[],"change_id":"I3be107406d57545db4e49bb705b82c3666b06436","subject":"fix: check for password_ref before assigning password expiry","status":"NEW","created":"2021-10-22 22:47:44.000000000","updated":"2021-12-08 16:19:57.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"229993d8b60962a05af4e18426a22cfc4de41f6a","_number":815168,"virtual_id_number":815168,"owner":{"_account_id":33826,"name":"Seena Fallah","email":"seenafallah@gmail.com","username":"clwluvw"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-10-23 00:55:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-23 00:46:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-23 00:55:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-18 11:16:03.000000000","updated_by":{"_account_id":33826,"name":"Seena Fallah","email":"seenafallah@gmail.com","username":"clwluvw"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"}],"messages":[{"id":"f21754c53bcef1e88eb1d10fc0c2cf4925ce14a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33826,"name":"Seena Fallah","email":"seenafallah@gmail.com","username":"clwluvw"},"date":"2021-10-22 22:47:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b3dae8100441b50ec7efe8ae852f31c9f1984eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-23 00:46:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/205491d9e210422fa8d7029c5644a33a : SUCCESS in 55m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8e5ab298b4dc4fb3a48601747e66614a : SUCCESS in 55m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bba5c85e353c7c8b44b14f38105440a14ccee65e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-23 00:55:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d136e59074ff4effade2eecfa5b762e0 : SUCCESS in 17m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07a38e8e3a454d1aa4f66c9746ea9f4c : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a4522fb86444232a4670e8dbd0c9b0b : SUCCESS in 14m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c09313c1648472f8436ef31c1aaf58c : SUCCESS in 15m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc761e73f31e48b8ae06b1c50065f51f : SUCCESS in 14m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/c99bacab7c7846e1b241ee88267dc058 : SUCCESS in 43m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bedcd33229014f5eb8c8724276783898 : SUCCESS in 1h 01m 00s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/794ee21e6ca54ab080128e61b71b35e5 : SUCCESS in 32m 56s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/26665ed13c0e445e92ca20ea56d56b40 : SUCCESS in 31m 01s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8c4fc964bf74a1b884aef1d928acb32 : SUCCESS in 35m 47s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/e7c11d6ba6de4800af60816cb5c23253 : SUCCESS in 24m 34s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/f51d803162aa495182a330cbc0f39e29 : SUCCESS in 21m 53s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/547144193880460485c608560277b6ec : FAILURE in 10m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9715bcb9b334abaa72fe5f58d7d7be5 : SUCCESS in 59m 21s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/08acd1a415474a6bae86956cc7f80ead : SUCCESS in 31m 47s","accounts_in_message":[],"_revision_number":1},{"id":"229993d8b60962a05af4e18426a22cfc4de41f6a","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-12-08 16:19:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f50c7cc6a562e5a03d19975eb05f8db24800f3df","revisions":{"f50c7cc6a562e5a03d19975eb05f8db24800f3df":{"kind":"REWORK","_number":1,"created":"2021-10-22 22:47:44.000000000","uploader":{"_account_id":33826,"name":"Seena Fallah","email":"seenafallah@gmail.com","username":"clwluvw"},"ref":"refs/changes/68/815168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/68/815168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/68/815168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/68/815168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/68/815168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/68/815168/1"}}},"commit":{"parents":[{"commit":"1e7ecca881a51144d61ae8026e1a77d6669997e2","subject":"Merge \"Fix typo in identity provider policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1e7ecca881a51144d61ae8026e1a77d6669997e2"}]}],"author":{"name":"Seena Fallah","email":"seenafallah@gmail.com","date":"2021-10-22 22:44:25.000000000","tz":210},"committer":{"name":"Seena Fallah","email":"seenafallah@gmail.com","date":"2021-10-22 22:47:39.000000000","tz":210},"subject":"fix: check for password_ref before assigning password expiry","message":"fix: check for password_ref before assigning password expiry\n\nIn case of creating user without password, setting password expiry should be done after checking password existence\n\nSigned-off-by: Seena Fallah \u003cseenafallah@gmail.com\u003e\nChange-Id: I3be107406d57545db4e49bb705b82c3666b06436\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f50c7cc6a562e5a03d19975eb05f8db24800f3df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f50c7cc6a562e5a03d19975eb05f8db24800f3df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
