)]}'
{"id":"openstack%2Fkeystone~821512","triplet_id":"openstack%2Fkeystone~stable%2Fxena~I69f16ed55dd9852859307b701a8391ba1e71c042","project":"openstack/keystone","branch":"stable/xena","topic":"bug/1897280-stable/xena","attention_set":{},"removed_from_attention_set":{"25666":{"account":{"_account_id":25666,"name":"Stuart Grace","email":"stuart.grace@bbc.co.uk","username":"stuartgrace"},"last_update":"2024-03-26 18:35:14.000000000","reason":"Change was abandoned"},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2024-03-26 18:35:14.000000000","reason":"Change was abandoned"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2024-03-26 18:35:14.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I69f16ed55dd9852859307b701a8391ba1e71c042","subject":"Accept STS and IAM services from Ceph Obj Gateway","status":"ABANDONED","created":"2021-12-13 09:15:16.000000000","updated":"2024-03-26 18:35:14.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":754404,"cherry_pick_of_patch_set":4,"meta_rev_id":"1c7e184a7f36f185d5ddf03f22cd7ee8ab9c07e7","_number":821512,"virtual_id_number":821512,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":25666,"name":"Stuart Grace","email":"stuart.grace@bbc.co.uk","username":"stuartgrace"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-07-15 17:59:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25666,"name":"Stuart Grace","email":"stuart.grace@bbc.co.uk","username":"stuartgrace"},{"value":1,"date":"2022-07-15 16:30:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2022-07-15 16:30:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":25666,"name":"Stuart Grace","email":"stuart.grace@bbc.co.uk","username":"stuartgrace"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25666,"name":"Stuart Grace","email":"stuart.grace@bbc.co.uk","username":"stuartgrace"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-13 09:15:16.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25666,"name":"Stuart Grace","email":"stuart.grace@bbc.co.uk","username":"stuartgrace"},"state":"REVIEWER"},{"updated":"2021-12-13 10:19:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-13 10:52:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-14 09:29:53.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"CC"},{"updated":"2022-01-14 18:22:17.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2022-01-14 18:22:46.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2022-01-14 18:23:01.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2022-05-13 04:38:49.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REMOVED"},{"updated":"2024-03-26 18:35:14.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"15d8785821bf348ab04a4608a3a8f307e2fc9847","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-12-13 09:15:16.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5fffecd7a57e8c30f58dfa93220e5726ec4f70e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 10:19:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/87f1c85676c34bc1ab8ce0bf4ad69fbf : TIMED_OUT in 1h 01m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a90aba77e5154b07ab61ed6d60dd66a8 : SUCCESS in 38m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e5a708cc8ea2c8b4ae52929ebbb599cf5a5fd81c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 10:52:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59d02f033f4c441f97baae41e0fe0b4c : SUCCESS in 42m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7611144494d24ea88f5edabf9eec3b86 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b909020d551b44b18de4af5a05c585d9 : TIMED_OUT in 41m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90a0de65d9124b63a706944fb69f213e : SUCCESS in 37m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/654d1cf8715b42478e01385544b9cac5 : SUCCESS in 15m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/09bba31992d641a3853d700d87513b42 : SUCCESS in 1h 13m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e3ac5f04e6844a2c89919df17995ac74 : SUCCESS in 1h 32m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e22d62a7ae2452091bf1208378199d9 : SUCCESS in 7m 58s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/72b6a66586a24074a4a924b70ffb1aec : SUCCESS in 27m 49s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/0f8adac739584f05b27bce62793f8b82 : SUCCESS in 40m 26s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bd3789c9ea24d25b2d9412325a52a59 : SUCCESS in 25m 44s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/be891f74f2f2435eb4d5fcc6d730ae95 : SUCCESS in 27m 51s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/8313edf4a930431f98e443ce7970899f : SUCCESS in 13m 56s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/9b60bad0c7024e0f87cdaefc69736bed : FAILURE in 12m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0149d70b2b8a4a288a0574eccf55a40d : SUCCESS in 42m 47s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/85b65dd562ad476bae0678dffb808630 : SUCCESS in 41m 53s","accounts_in_message":[],"_revision_number":1},{"id":"fbaf4b3a5178a9241021d9f36443c006afae140f","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-12-14 09:29:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3093501a6c468eda3fd9855560de71eec4b5f3be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 10:26:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d061255eda249e0837b2f40d0775060 : SUCCESS in 19m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8de7b4c67b7a43569a74d4bea7aa79d2 : SUCCESS in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4365b87b01cd4e38a71e098fafcfc082 : SUCCESS in 13m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/261429ddb4c34c009691e1f4e8bd635b : SUCCESS in 29m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f84fa2a977d848f094e9b57f26aacdf8 : SUCCESS in 13m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/42122aa2855f4b40bd8c0d7c1ce35cf3 : FAILURE in 30m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9a60e5776d314b709418831379611841 : FAILURE in 29m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75b3e50f96964b3f9e6b3ec2fa6b3ed2 : SUCCESS in 7m 53s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/6d8d69cdf5c44b1db02c1981e85324d9 : FAILURE in 30m 57s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/7ef0d8f1b7af4846a7a0192e175d3597 : FAILURE in 34m 08s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f9323c8417b4de1b0ffc3d8a10cd63e : FAILURE in 27m 00s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/9373e1fc5d0b408f8a887588ec4ccc79 : SUCCESS in 32m 34s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/7dbece8ddcfb42c48da0cba92429af76 : SUCCESS in 14m 29s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/20c4c0a1b5c84494a334a241a37c1f5b : FAILURE in 14m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f40dff0458814fa6a9101c409f75d0fc : SUCCESS in 52m 12s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/c59577404c184794a08526689b59ba97 : FAILURE in 23m 48s","accounts_in_message":[],"_revision_number":1},{"id":"16c7a9142592cde5ebe2c4d1deffb7a15e3aef05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 10:34:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/61e1cbaefe5f44a092c6cccf444f57e7 : TIMED_OUT in 1h 01m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d33eb30f1b3b4900892a8d1af0cdb226 : SUCCESS in 39m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bec383570c73c7f78c4c6b747414d5f3fbee076c","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-01-12 08:52:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2bd4300b98a18e92939f2ae40ada92f0d9cd56cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 10:06:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac034839a3c3488d8cca51982424a601 : SUCCESS in 58m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e542d408a6b489fa08722893caa35e2 : SUCCESS in 53m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"edbca72dacf451badb7c64d07a32abd300a7e47a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 16:56:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48603030aa814f85baff12ae061aa635 : SUCCESS in 24m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b684654f5ca4c1f8c2c5ef2b76c1dfc : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb3d6e6911954c0eae2d102a14d553be : SUCCESS in 14m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8540f6537584330942214cc117de2c4 : SUCCESS in 25m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f5e25fd13d64e0792f53b5a8aa1225f : SUCCESS in 14m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/5f41b76be5564ae7b30c416b499b0138 : SUCCESS in 1h 07m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1e225d96290a49a488d056accf51fbc8 : SUCCESS in 1h 37m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf41e6327a9f405f8a6a909b9669e4da : SUCCESS in 9m 34s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f897e002729247f6ac38e50de819463c : SUCCESS in 30m 17s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/3fbe155edbb1456aa35b561648e60f16 : SUCCESS in 45m 21s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eaf211efec1b4b8c91a86c41ba3fd3cc : SUCCESS in 35m 32s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/5b07f8a75803468d8b193ffa16f00538 : SUCCESS in 34m 34s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/8df7447d00f14c13bf1d715920a4e6b9 : SUCCESS in 19m 52s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/2d7c1f5b32c84c499f8167235b1ff80b : FAILURE in 14m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/360f001053c54a949124d28de2fe3598 : SUCCESS in 49m 03s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/0b42a193442b42b4b5605400d0403462 : SUCCESS in 46m 25s","accounts_in_message":[],"_revision_number":1},{"id":"c1392f6239e1593a3f8945dca7da21964882c46a","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-01-14 18:22:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f06b52f5000d9b976499c49f47451fa1a475efdf","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-01-14 18:22:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1ec97c75996ebbf2d3df6aefc89793b3775ace9d","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-01-14 18:23:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"159d96b31c6ba005dc08e51ea386c7da65d2cb38","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-05-13 04:38:49.000000000","message":"Removed reviewer Gage Hugo with the following votes:\n\n* Code-Review+1 by Gage Hugo \u003cgagehugo@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"b09ff05995fce48400df617b0bd6ab9e23b00a61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-07-15 16:30:16.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e077d5f66c2983528a4729b0994b839524726788","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 17:19:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc8a6644b64c4f3faa95516b8ff2d34f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/08065cfb2a464e87a7fc3bec4b1dff55 : SUCCESS in 39m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff723f742eb146eb81b0b2f53efdcbb2 : SUCCESS in 38m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7b7a892dd1ae799c24e38329c511aecc58fbb179","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 17:59:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c08e46dac2744a45bb960ab8dd7f429a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d16019eab1ab41498fb9d78b536915bf : SUCCESS in 19m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42070b6438574483a9171930eabae0e6 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c374277dc98494cb4fec971e671eb60 : SUCCESS in 22m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b73b6033cafe496f96d777ab751bba9f : SUCCESS in 21m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c49488a3cd704cb69054d00f281c180d : SUCCESS in 15m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/994229ef983a4201a35f8e6c635409f0 : SUCCESS in 15m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/4023d0acfb0f4697ad486b8ff505654e : SUCCESS in 1h 01m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5300ba4618a4aac80a71798e6133131 : SUCCESS in 1h 27m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a4e2a3b38c5448f91ccea0a505a9892 : SUCCESS in 9m 35s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/7a853cfc08d947e4aec5aa17ea9470ec : SUCCESS in 40m 50s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/c6efde953c2e45b7979d82942c50d569 : RETRY_LIMIT in 12m 33s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0aebfa9599d641ef8af9ea26ab80ea3d : SUCCESS in 37m 28s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/c808ae19cf774e7f9bacd6ccd44807d8 : SUCCESS in 32m 57s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/21e3edc1663448359bebe1690792b785 : SUCCESS in 22m 21s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/b1796a4a7fca4f2b9bf4be157d3ae00e : FAILURE in 16m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cbd968b5dd044e59949d8ad950f9f5ff : SUCCESS in 59m 27s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/e9e7bfec2ef04fcb832621fba90ff334 : SUCCESS in 45m 32s","accounts_in_message":[],"_revision_number":2},{"id":"1c7e184a7f36f185d5ddf03f22cd7ee8ab9c07e7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-26 18:35:14.000000000","message":"Abandoned\n\nstable/xena branch of openstack/keystone is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/xena if you want to further work on this patch.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9680ce52f622ca7c5b575d9677286ad216478e3b","revisions":{"b44b60c183aa86fae06c70ade429e9b10fd1e345":{"kind":"REWORK","_number":1,"created":"2021-12-13 09:15:16.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/12/821512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/12/821512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/821512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/821512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/821512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/12/821512/1"}}},"commit":{"parents":[{"commit":"20be021347664438520f420eb26168435f7b860b","subject":"Merge \"Fix typos in ec2 credential policies\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/20be021347664438520f420eb26168435f7b860b"}]}],"author":{"name":"Stuart Grace","email":"stuart.grace@bbc.co.uk","date":"2020-09-25 14:10:22.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-12-13 09:15:16.000000000","tz":0},"subject":"Accept STS and IAM services from Ceph Obj Gateway","message":"Accept STS and IAM services from Ceph Obj Gateway\n\nCeph Object Gateway can use keystone for authenticating user requests\nto its S3-compatible API, but recent versions also provide two other\nAWS-compatible APIs for managing user access: Security Token Service\n(STS) and Identity and Access Management (IAM). These attempt to\nauthenticate requests with Keystone but always receive 403 Access\nDenied because _calculate_signature_v4() in api/s3tokens.py only\naccepts \"s3\" as the service name. This patch accepts any of \"s3\" or\n\"sts\" or \"iam\" as valid service names.\n\nChange-Id: I69f16ed55dd9852859307b701a8391ba1e71c042\nCloses-Bug: #1897280\n(cherry picked from commit 36d6fc7f8f06b4b4d1af09bd56b3c17fa1a502ca)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b44b60c183aa86fae06c70ade429e9b10fd1e345"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b44b60c183aa86fae06c70ade429e9b10fd1e345"}]},"branch":"refs/heads/stable/xena"},"9680ce52f622ca7c5b575d9677286ad216478e3b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-07-15 16:30:16.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/12/821512/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/12/821512/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/821512/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/821512/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/821512/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/12/821512/2"}}},"commit":{"parents":[{"commit":"bd692531e98ce962c892059f32c2ba37ce3291e3","subject":"Xena-only: Fix wrong python job template used","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bd692531e98ce962c892059f32c2ba37ce3291e3"}]}],"author":{"name":"Stuart Grace","email":"stuart.grace@bbc.co.uk","date":"2020-09-25 14:10:22.000000000","tz":60},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-07-15 16:30:16.000000000","tz":0},"subject":"Accept STS and IAM services from Ceph Obj Gateway","message":"Accept STS and IAM services from Ceph Obj Gateway\n\nCeph Object Gateway can use keystone for authenticating user requests\nto its S3-compatible API, but recent versions also provide two other\nAWS-compatible APIs for managing user access: Security Token Service\n(STS) and Identity and Access Management (IAM). These attempt to\nauthenticate requests with Keystone but always receive 403 Access\nDenied because _calculate_signature_v4() in api/s3tokens.py only\naccepts \"s3\" as the service name. This patch accepts any of \"s3\" or\n\"sts\" or \"iam\" as valid service names.\n\nChange-Id: I69f16ed55dd9852859307b701a8391ba1e71c042\nCloses-Bug: #1897280\n(cherry picked from commit 36d6fc7f8f06b4b4d1af09bd56b3c17fa1a502ca)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9680ce52f622ca7c5b575d9677286ad216478e3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9680ce52f622ca7c5b575d9677286ad216478e3b"}]},"branch":"refs/heads/stable/xena","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
