)]}'
{"id":"openstack%2Fkeystone~823551","triplet_id":"openstack%2Fkeystone~master~I7b0058961e3a48a880c73b05113efb17de99356d","project":"openstack/keystone","branch":"master","topic":"yourijing","attention_set":{"16465":{"account":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"last_update":"2022-01-14 18:55:24.000000000","reason":"Douglas Mendizábal replied on the change"},"6593":{"account":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"last_update":"2022-01-06 04:01:59.000000000","reason":"yourijing replied on the change"},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2022-01-14 18:55:24.000000000","reason":"Douglas Mendizábal replied on the change"},"34392":{"account":{"_account_id":34392,"name":"bieji","display_name":"yourijing","email":"linjiangbieji@qq.com","username":"linjiang"},"last_update":"2022-01-14 19:01:04.000000000","reason":"Gage Hugo replied on the change"}},"removed_from_attention_set":{"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2022-05-27 15:43:33.000000000","reason":"David Wilde replied on the change"},"21420":{"account":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"last_update":"2022-01-14 19:01:04.000000000","reason":"Gage Hugo replied on the change"}},"hashtags":[],"change_id":"I7b0058961e3a48a880c73b05113efb17de99356d","subject":"fix error, if two password create at same time, maybe not chose right","status":"NEW","created":"2022-01-05 15:11:17.000000000","updated":"2022-05-27 15:44:26.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"f0f25051a32a05f5e0843a002cda68218f5f111d","_number":823551,"virtual_id_number":823551,"owner":{"_account_id":34392,"name":"bieji","display_name":"yourijing","email":"linjiangbieji@qq.com","username":"linjiang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-06 01:38:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":-1,"date":"2022-05-27 15:43:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-05 16:15:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-05 17:00:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-06 04:01:59.000000000","updated_by":{"_account_id":34392,"name":"bieji","display_name":"yourijing","email":"linjiangbieji@qq.com","username":"linjiang"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2022-01-06 04:01:59.000000000","updated_by":{"_account_id":34392,"name":"bieji","display_name":"yourijing","email":"linjiangbieji@qq.com","username":"linjiang"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2022-01-14 18:55:24.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2022-01-14 18:55:24.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2022-01-14 18:55:24.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2022-05-13 04:38:33.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REMOVED"},{"updated":"2022-05-27 15:44:26.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"CC"}],"messages":[{"id":"fcba6bd683f87bc01ec8b1fe70ed738c7c66df98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34392,"name":"bieji","display_name":"yourijing","email":"linjiangbieji@qq.com","username":"linjiang"},"date":"2022-01-05 15:11:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b51e6729ec7ed8fdf10365af8bb36eae7fca86a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 16:15:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a5bb5aaaf974567a0a673c521ecad05 : SUCCESS in 38m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4b1470cf3c0d4780a5146276123a1092 : TIMED_OUT in 1h 01m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8e83c4dfdbd195d68d0c5bd2281d2d5aa4cf44b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 17:00:15.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/058afdc1f6bb4fe0b7e154952dcb7549 : SUCCESS in 19m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61f1c11194fc4624a3e26fbfaed35033 : FAILURE in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecf29f3e55fb4b6aaa17d1125bb74a18 : SUCCESS in 14m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5d7294a4f3346139f5121265444b19f : SUCCESS in 13m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/250fee00679a47b892d49b93b7459522 : SUCCESS in 16m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/a1b95a6ca9e74fe3bedaaca4d029b1f1 : SUCCESS in 1h 28m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14e51e30aaae4d39b19fc19e2d0c408a : SUCCESS in 1h 40m 40s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/7ce366e329fa4e4a90c7d526de91e3a0 : SUCCESS in 32m 25s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/74cf4ae819674fc08fb1cad3a9c2228b : SUCCESS in 58m 06s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5205b2e4e2b748b0981b3365192142bc : SUCCESS in 32m 47s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/6b1eb8db4431479a8c0a512576aa55a9 : SUCCESS in 37m 47s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/712aacead1554f94ad49d64ad9978cd9 : SUCCESS in 18m 19s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ed7ff6f39eb44803907c8cf9b45ac1cf : SUCCESS in 46m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4297509d757a4ed18ef9547f177fc575 : SUCCESS in 53m 25s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/5f67f53a255e4da3a8fa5773c8f00955 : SUCCESS in 54m 25s","accounts_in_message":[],"_revision_number":1},{"id":"75754d528fe6f62dee9e2e04a6601b83315e5abc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34392,"name":"bieji","display_name":"yourijing","email":"linjiangbieji@qq.com","username":"linjiang"},"date":"2022-01-06 00:06:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b15695d54d2253d752c42a3129557ef30209cf56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 01:10:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e046ae5dcee44cefbea98422cba767a2 : SUCCESS in 54m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e76704a5496f41bd9ed95c3cb4744525 : TIMED_OUT in 1h 01m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4f02b835cebaaba361fe0e08cec60030269b9d28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 01:38:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eaa39e79cd1149a6b3bda3725763faad : SUCCESS in 25m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7838063222f46048737001b782ab0b5 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ab3f6724e9b4e5599eed2666bd424b4 : SUCCESS in 19m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f26c9af7727046e5bf338e27f2d48e6e : SUCCESS in 16m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a602dfd12b884faebe18630b5c76b994 : SUCCESS in 15m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/97c037405c224b8bbc7cf60bab5d3cc9 : SUCCESS in 1h 00m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe79e92231f2465dbb1cf94c9c69ea33 : SUCCESS in 1h 29m 52s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/a1d49455bbc149c4bd94b53f92ba2470 : SUCCESS in 28m 50s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/74c9b1dbc2aa4d9087d479c95295ce58 : SUCCESS in 36m 41s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd6110b635f64d15ae9b2d1829b03d44 : FAILURE in 33m 04s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/f6390418b6844418a31812741c59cffc : SUCCESS in 28m 33s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/afc14920389148ec9db9e0c0135a5787 : SUCCESS in 14m 51s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/e6724e9cc7fa4d7996f99229bdac5fcd : SUCCESS in 35m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c056169057cd41b294e6b31b5c8faecf : SUCCESS in 47m 57s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/2e0e355ba9024768b1396cf198e98364 : SUCCESS in 38m 54s","accounts_in_message":[],"_revision_number":2},{"id":"e9cd1a4efa25eab7f4493af6742fb88673afae3d","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-01-14 19:01:04.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b51c57937c9241179350341f69cb3eb685e10496","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-05-13 04:38:33.000000000","message":"Removed reviewer Gage Hugo with the following votes:\n\n* Code-Review-1 by Gage Hugo \u003cgagehugo@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"dde4c794e9f2fcaa9e4f754819e3434d3b75d0ca","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-05-27 15:43:33.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f0f25051a32a05f5e0843a002cda68218f5f111d","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2022-05-27 15:44:26.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d9ebab7ef955e4c34ebc63c8f37b7d6940febb77","revisions":{"14441d62aee86c5c233fe40c52afd5e7fd275ad1":{"kind":"REWORK","_number":1,"created":"2022-01-05 15:11:17.000000000","uploader":{"_account_id":34392,"name":"bieji","display_name":"yourijing","email":"linjiangbieji@qq.com","username":"linjiang"},"ref":"refs/changes/51/823551/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/51/823551/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/823551/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/823551/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/823551/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/51/823551/1"}}},"commit":{"parents":[{"commit":"a9fa5131cda296743e4916577ed968b488a414d6","subject":"Merge \"Replace assertItemsEqual with assertCountEqual\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a9fa5131cda296743e4916577ed968b488a414d6"}]}],"author":{"name":"linjiang","email":"linjiangbieji@qq.com","date":"2022-01-05 15:08:52.000000000","tz":480},"committer":{"name":"linjiang","email":"linjiangbieji@qq.com","date":"2022-01-05 15:08:52.000000000","tz":480},"subject":"fix error, if two password create at same time, maybe not chose right","message":"fix error, if two password create at same time, maybe not chose right\n\nChange-Id: I7b0058961e3a48a880c73b05113efb17de99356d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/14441d62aee86c5c233fe40c52afd5e7fd275ad1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/14441d62aee86c5c233fe40c52afd5e7fd275ad1"}]},"branch":"refs/heads/master"},"d9ebab7ef955e4c34ebc63c8f37b7d6940febb77":{"kind":"REWORK","_number":2,"created":"2022-01-06 00:06:40.000000000","uploader":{"_account_id":34392,"name":"bieji","display_name":"yourijing","email":"linjiangbieji@qq.com","username":"linjiang"},"ref":"refs/changes/51/823551/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/51/823551/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/823551/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/823551/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/823551/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/51/823551/2"}}},"commit":{"parents":[{"commit":"a9fa5131cda296743e4916577ed968b488a414d6","subject":"Merge \"Replace assertItemsEqual with assertCountEqual\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a9fa5131cda296743e4916577ed968b488a414d6"}]}],"author":{"name":"linjiang","email":"linjiangbieji@qq.com","date":"2022-01-05 15:08:52.000000000","tz":480},"committer":{"name":"linjiang","email":"linjiangbieji@qq.com","date":"2022-01-06 00:06:02.000000000","tz":480},"subject":"fix error, if two password create at same time, maybe not chose right","message":"fix error, if two password create at same time, maybe not chose right\n\nChange-Id: I7b0058961e3a48a880c73b05113efb17de99356d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d9ebab7ef955e4c34ebc63c8f37b7d6940febb77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d9ebab7ef955e4c34ebc63c8f37b7d6940febb77"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
