)]}'
{"id":"openstack%2Fkeystone~823659","triplet_id":"openstack%2Fkeystone~master~Ia2046dc32e3ac270b1dbcf2fe540104c1a8d95d8","project":"openstack/keystone","branch":"master","topic":"bp/remove-sqlalchemy-migrate","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-01-10 22:13:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia2046dc32e3ac270b1dbcf2fe540104c1a8d95d8","subject":"Add \u0027WarningsFixture\u0027","status":"MERGED","created":"2022-01-06 10:48:49.000000000","updated":"2022-01-10 22:18:26.000000000","submitted":"2022-01-10 22:13:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823659-bp/remove-sqlalchemy-migrate","meta_rev_id":"c9316d5e021908ddac10f3548a7feb6f816ddfec","_number":823659,"virtual_id_number":823659,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-10 22:13:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2022-01-10 15:01:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2022-01-06 15:45:05.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2022-01-10 15:01:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-06 11:43:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-06 14:55:32.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"CC"},{"updated":"2022-01-06 15:45:05.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2022-01-10 15:01:00.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"}],"messages":[{"id":"3e9d9352d5b7d427d7fa1d6d8c18aac5c6cae8ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-06 10:48:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7305deabb5e3adbe11a8781a6f569c27fcd281a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 11:43:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cea4996b6d124a6eb16cdc5ccda77ebe : SUCCESS in 19m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cc9a4dff4a14ede9a4fe94b30165eea : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52a56563b1fe4dc78328cb438183672e : SUCCESS in 14m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adabe45c698d4bb9bdf7a3aec88274c0 : SUCCESS in 15m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3034a5f24504de18391e2392aec886d : SUCCESS in 16m 17s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/c597174d7eed4ae09f0f151bcb3e830d : FAILURE in 50m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d0a4d0b2991cc7a9087a78b3aa8377b19b079bcf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 12:11:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6af8fe7453cc4045a6aa97d8fe7bf784 : SUCCESS in 49m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6cfe041ec36c4774a09d774c9c5cf7e6 : TIMED_OUT in 1h 02m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e72a6d8302241b6d65a1159fb93e0bef37495f3","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-01-06 14:55:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b17edade440bd24cf082085b37061be2180ddef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 15:42:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd132ae319554fb4b10c16312e2b998a : SUCCESS in 41m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8460332d57db446195ec1c94e0ec41b9 : SUCCESS in 42m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f1e7f69fe46115d1f146b67444eddff535d402a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 15:43:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0012102766742559fd68b707710f992 : SUCCESS in 22m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9556eac30214f7e84974a5094d0f416 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4435f7fcede84584993ed5f2c539dd06 : SUCCESS in 16m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ddfd77d289f84acd8cb80be8a76a2857 : SUCCESS in 17m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35993e8cf3404975aebd4d59760eccbb : SUCCESS in 15m 23s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/074fe1a41ed84a8aa84e5cf69137ee51 : SUCCESS in 44m 20s","accounts_in_message":[],"_revision_number":1},{"id":"379756f5584fae254c4f7b09f12474aa6d9e935e","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-01-06 15:45:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"916cd110bc47af2ab6dd16ae134b0366c13dc006","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-01-10 15:01:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"767173f7bb79c8f03f97e6f0e6666accfdf8cdaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 15:02:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4a96cf53c3672329218ce5dfbb1dfd4023aea225","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 22:13:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4deb12ee6037432695caf471618b3aa4 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32f3526b2826417b953c81c83818a802 : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f730ce10ac9945ce858829500a444f3e : SUCCESS in 13m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5686f1afb39c4f6db4b04b14822c8025 : SUCCESS in 15m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/840c5666356f4f5683cf4210de2383c1 : SUCCESS in 16m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/c46a5fef408c4494ab5cec1ba9733c7f : SUCCESS in 1h 03m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55c8283e7d364e808c75f4b2afe05927 : SUCCESS in 1h 27m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/002f0222be2b473b9c7ad3e579c9dabd : SUCCESS in 9m 23s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/7309182305b44227b772136371976a5a : SUCCESS in 27m 58s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/d0568482aefd4ad4a14bdac430a4372a : SUCCESS in 35m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8f7e56dc54c4740bc7011551b894179 : SUCCESS in 1h 03m 37s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/d6f2f789a679411b8b2f34c6c98244f5 : SUCCESS in 46m 56s","accounts_in_message":[],"_revision_number":1},{"id":"792e5dd81afdf430d59493a02d9829c811e71385","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 22:13:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c9316d5e021908ddac10f3548a7feb6f816ddfec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 22:18:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/994fd011809848aeb7574aa312661dca : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"771c943ad2116193e7bb118c74993c829d93bd71","revisions":{"771c943ad2116193e7bb118c74993c829d93bd71":{"kind":"REWORK","_number":1,"created":"2022-01-06 10:48:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/59/823659/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/59/823659/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/59/823659/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/59/823659/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/59/823659/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/59/823659/1"}}},"commit":{"parents":[{"commit":"03238e343a595e1cb4fd43a84062db72b094c832","subject":"Add support for pysaml2 \u003e\u003d 7.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/03238e343a595e1cb4fd43a84062db72b094c832"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-12-24 11:24:08.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-05 18:29:43.000000000","tz":0},"subject":"Add \u0027WarningsFixture\u0027","message":"Add \u0027WarningsFixture\u0027\n\nThis duplicates what exists in nova and various other projects. The\nimportant difference between this and what we\u0027re doing currently is that\nit *restores*, rather than reset, the warning filters. There are more\nvarious warning filters pre-configured in a typical Python environment,\nincluding a few from third-party libraries such as requests [1][2] and\nurllib3 [3] as well as stdlib [4]. By calling \u0027warnings.resetwarnings\u0027, we\n*reset* all the warning filters [5]. This is clearly not something we want to\ndo, and resulted in tests puking warnings after the initial test run.\n\n[1] https://github.com/psf/requests/blob/v2.26.0/requests/__init__.py#L127\n[2] https://github.com/psf/requests/blob/v2.26.0/requests/__init__.py#L152\n[3] https://github.com/urllib3/urllib3/blob/1.26.7/src/urllib3/__init__.py#L68-L78\n[4] https://docs.python.org/3.8/library/warnings.html#default-warning-filter\n[5] https://docs.python.org/3.8/library/warnings.html#warnings.resetwarnings\n\nChange-Id: Ia2046dc32e3ac270b1dbcf2fe540104c1a8d95d8\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/771c943ad2116193e7bb118c74993c829d93bd71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/771c943ad2116193e7bb118c74993c829d93bd71"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
