)]}'
{"id":"openstack%2Fkeystone~823661","triplet_id":"openstack%2Fkeystone~master~I161454bb5d112abf32e8ae363955b36a7529889d","project":"openstack/keystone","branch":"master","topic":"bp/remove-sqlalchemy-migrate","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-01-19 18:50:47.000000000","reason":"stephenfin replied on the change"}},"hashtags":[],"change_id":"I161454bb5d112abf32e8ae363955b36a7529889d","subject":"sql: Squash liberty migrations","status":"MERGED","created":"2022-01-06 10:48:49.000000000","updated":"2022-01-19 23:56:53.000000000","submitted":"2022-01-19 23:49:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823661-bp/remove-sqlalchemy-migrate","meta_rev_id":"401ac3f58567c6f746304b13a80248aeba7e4a8d","_number":823661,"virtual_id_number":823661,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-19 23:49:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2022-01-18 19:06:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2022-01-14 18:38:49.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2022-01-14 18:41:08.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2022-01-14 18:43:45.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2022-01-18 19:06:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-06 12:11:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-06 12:39:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-14 18:38:49.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2022-01-14 18:41:08.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2022-01-14 18:43:45.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2022-01-18 19:06:13.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"}],"messages":[{"id":"0aee27887329c4d567481c509d9d829ddd7af2e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-06 10:48:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a774a1003af763f616ee8cba8d821fdd442fb10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 12:11:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/471547e628034d18a35c2689970408ae : TIMED_OUT in 1h 02m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ae4d7cb164a499480da4685291ceecc : TIMED_OUT in 1h 02m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"56aa6c72d1ec0cd896a48a87b749e1f8e551d783","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 12:39:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f5d919bcbab4e119a113dd4d54a0228 : SUCCESS in 20m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0736d3913cbf4888b4edc952d53f35de : SUCCESS in 9m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f4d2a7d96f3491aac233afb880db1e8 : SUCCESS in 24m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/607e3ef993ac4887b97d314c1795f19c : SUCCESS in 26m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7904f350ef6a4727a1ec0b44c0a92e61 : SUCCESS in 14m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/69606c2fd68244219c659509c7cedf90 : SUCCESS in 1h 06m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7d5ce8b6304e422c8f0efe088dcdc8e0 : SUCCESS in 1h 33m 52s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f0d0f1fa79064b7489a19dafcc654c3b : SUCCESS in 36m 20s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/222ce0c867db4b3b92e751e1888ae23e : SUCCESS in 46m 12s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bdad7d17922405ebef01fe47d60e3f4 : SUCCESS in 38m 47s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/b0de0bf88c2743a3b3df154ff1075d01 : SUCCESS in 33m 05s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d3062d61d7ec4a8990fe44e5978d94c1 : SUCCESS in 19m 05s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/28882885f0fa483199523583d4039e8b : SUCCESS in 37m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c36f282a54f4523bb860c20d1873533 : SUCCESS in 52m 15s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/e3d89389294446d095c3e62c3c5ddbdf : SUCCESS in 42m 11s","accounts_in_message":[],"_revision_number":1},{"id":"4bbcd319181ad9477c673fa754501010421f782f","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-01-14 18:38:49.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3232d19009f4b1dbda746a021c7b6bbe264c0971","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-01-14 18:41:08.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5f409cab074460f11b1f17d9821617e1d2c02ef1","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-01-14 18:43:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f9fe6e13abd2d8f7fa9dc1e6ce7ec9a26a076c09","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-01-18 19:06:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"adac6c55483e0b74fd0459bffcdd31841d49fb80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 19:07:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b7f92e4a22e098c1faa5cb4d9206d727a138158a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 20:48:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01b52a3b8ad645a3a048d684d1f7e9d6 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4387dc438c94ee38756de48348f2af9 : SUCCESS in 16m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2ac5d8af72247af9ed393e9e0179d52 : SUCCESS in 15m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdd012502dfe4fb3b330699dc03f6e51 : SUCCESS in 14m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/fcd015146e364b5e8173dac894c3f7e2 : SUCCESS in 57m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c6b37173494e467b89242192cd1c2735 : SUCCESS in 1h 34m 58s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f981583ddcdc4fadba2ce56a87075058 : SUCCESS in 28m 20s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/8427470f4c2044f7b6afcf37cd7cf75d : SUCCESS in 32m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6e2bfd3a58447b5994d9782a88a3245 : SUCCESS in 52m 42s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/6623a8934a24475c89d483e04e29443d : FAILURE in 41m 19s","accounts_in_message":[],"_revision_number":1},{"id":"26abca8b3e66b770d285de3cc9c1b23d16f133d9","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-01-19 16:13:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6ae8415e4f8089efcccfe80a0aeb94ba5acd06e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 16:58:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/db69b93094bb4f9b82f4c1de0f0136e3 : SUCCESS in 37m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/71d1d88f66044b18ab3a207ffbf495f8 : SUCCESS in 38m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"70a368af4a6062c286416d9d997fcc0fe3296915","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 17:24:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1756061aa9a4338a4ad427bc008647c : SUCCESS in 32m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e79adfbebe974293bc9dbd3e830232bc : SUCCESS in 8m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/999e089fd0de4b71a49018b920bbae4c : SUCCESS in 26m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b40c168e22e7429ab315ff20cf941b5b : SUCCESS in 16m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f89ba4d732e4ddf8c51640ee1cf6618 : SUCCESS in 14m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/12a4ad35794e425fae3a2fbc559411cf : SUCCESS in 47m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9cd7b8665fce48c889c8650a0cb879aa : SUCCESS in 1h 07m 01s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/03d1236ce32544168d808a1c066138fd : SUCCESS in 21m 52s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/3c90c55a9bbb4e16bcefa0f1d60471da : SUCCESS in 31m 37s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/221e56c93cdd461aac69fd29d883e924 : SUCCESS in 31m 16s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/384802ef5d2f498c9272af31ff1c080f : SUCCESS in 31m 02s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/6248be9f83c1439785033875cb56a1d4 : SUCCESS in 13m 07s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/f00dd0c2c9eb404b9bdc0230bb2f2aa8 : SUCCESS in 49m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80ecf20a0d394e089c7cc071dd7e5bdd : SUCCESS in 47m 59s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/7e7d45b4852a48dfb96b45f8d6d01137 : FAILURE in 42m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b9c75776741146be4584bc4aa74ade23815d4d4e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-19 18:50:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6175b46644efc96fa7b952ca00726df77725ab94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 20:49:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd092015995b4b5898684609e73e18f5 : SUCCESS in 17m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bff8c32fba54555969c86116f80ca62 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfc3b764d6324aebac8b7d3dc6c4223c : SUCCESS in 13m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12f107370f0e4bb8ba97407407c187ba : SUCCESS in 15m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c1776ed360b4daf990842eb16ad4ad1 : SUCCESS in 14m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/61ce0b14c43c47cb8ce86f93d3b27961 : SUCCESS in 1h 06m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a0ad1a30c2234ee68947248e2f8f8cd2 : SUCCESS in 1h 37m 02s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/1a75fdf58e84468c939c68f3897f83ec : SUCCESS in 38m 48s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/4c19f08af2b94dfe894f5f68a2c518f9 : SUCCESS in 53m 01s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6112d2d8adfd45bcab147199a9088c21 : SUCCESS in 35m 37s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/11afc5b948e64c35b4d26050112f55fb : SUCCESS in 33m 57s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/68913d228ae34a978715d60339cafa19 : SUCCESS in 20m 00s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/7e38a142f30c4df9b033b30de1bcbebf : SUCCESS in 32m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1ac369098e54d8c999951beba07de4b : SUCCESS in 1h 02m 22s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/cdeb749eb5ee43f8a164088846b933dd : SUCCESS in 41m 45s","accounts_in_message":[],"_revision_number":1},{"id":"33e5b191a6ac60fa5a01acf1026b2914dd19f5a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 20:54:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e01a706bf74608065b4c5970af08155a813f37e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 21:07:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c1059ab60d1f478baba567c30d474228 : TIMED_OUT in 1h 01m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/35895b66a5c34154b4bc5909af7af990 : TIMED_OUT in 1h 01m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee6ebd2bb2039e11897f36ae8ac072c885584183","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 23:49:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b494bc2c9d4a47248ad788c3b0077269 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf6618684e8942288aedcbc7ee8569fa : SUCCESS in 15m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29a6048e538c4420a31a8eb880804a41 : SUCCESS in 16m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a29f203aabda4358b7e9dcfae3d6a4a0 : SUCCESS in 15m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/79c339363739468c868679aef3056e72 : SUCCESS in 1h 19m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7951e0890104ef6af941a61c841e25d : SUCCESS in 2h 05m 53s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/451d678193ca4224b96b4e85306e9031 : SUCCESS in 29m 04s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/a01cf119a5dc440ba1f3fe710dc3d908 : SUCCESS in 31m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0db77d711ee436b876582c51c1be839 : SUCCESS in 1h 00m 36s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/c486b39811b040188be8fe532a13b16e : SUCCESS in 43m 49s","accounts_in_message":[],"_revision_number":1},{"id":"4e5c496c957bb9c2a802d129d217ad85b81f7ef7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 23:49:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"401ac3f58567c6f746304b13a80248aeba7e4a8d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 23:56:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb731795ac444d61aaa722c4bb9d76bb : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c80b183aa5eb78b68e41edae4fccf8a8d831ded1","revisions":{"c80b183aa5eb78b68e41edae4fccf8a8d831ded1":{"kind":"REWORK","_number":1,"created":"2022-01-06 10:48:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/61/823661/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/61/823661/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/61/823661/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/61/823661/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/61/823661/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/61/823661/1"}}},"commit":{"parents":[{"commit":"e92bf89f71334b934d974165e9d142c684b9fc1c","subject":"sql: Trivial formatting changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e92bf89f71334b934d974165e9d142c684b9fc1c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-12-23 12:01:11.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-05 18:29:43.000000000","tz":0},"subject":"sql: Squash liberty migrations","message":"sql: Squash liberty migrations\n\nMake the following changes to the new \"initial\" migration.\n\n- Add \u0027inherited\u0027 column to primary key constraint for \u0027assignment\u0027\n  table (073)\n- Add \u0027is_domain\u0027 column to \u0027project\u0027 table (074)\n- Add \u0027config_register\u0027 table (075)\n\nSchemas can be compared using the techniques discussed at [1]. A sample\ncommand for SQLite:\n\n  # with change before this\n  python manage.py version_control \\\n    --database \u0027sqlite:///before.db\u0027 --repository . --version 066\n  python manage.py upgrade \\\n    --database \u0027sqlite:///before.db\u0027 --repository . --version 075\n\n  # with this change\n  python manage.py version_control \\\n    --database \u0027sqlite:///after.db\u0027 --repository . --version 074\n  python manage.py upgrade \\\n    --database \u0027sqlite:///after.db\u0027 --repository . --version 075\n\n[1] https://that.guru/blog/comparing-nova-db-migrations/\n\nChange-Id: I161454bb5d112abf32e8ae363955b36a7529889d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c80b183aa5eb78b68e41edae4fccf8a8d831ded1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c80b183aa5eb78b68e41edae4fccf8a8d831ded1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
