)]}'
{"id":"openstack%2Fkeystone~823663","triplet_id":"openstack%2Fkeystone~master~Ib6ecae7a2e6fa3780106fa6a4b50d47aab922b43","project":"openstack/keystone","branch":"master","topic":"bp/remove-sqlalchemy-migrate","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-01-20 09:37:30.000000000","reason":"stephenfin replied on the change"}},"hashtags":[],"change_id":"Ib6ecae7a2e6fa3780106fa6a4b50d47aab922b43","subject":"sql: Squash newton migrations (part 1)","status":"MERGED","created":"2022-01-06 10:48:49.000000000","updated":"2022-01-20 12:35:03.000000000","submitted":"2022-01-20 12:32:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823663-bp/remove-sqlalchemy-migrate","meta_rev_id":"8d73ea8de18241a890419e7ca6b9ff10b53ff9da","_number":823663,"virtual_id_number":823663,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-20 12:32:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2022-01-18 19:11:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-19 20:41:29.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2022-01-18 19:11:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-06 12:11:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-06 12:33:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-14 18:49:18.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2022-01-14 18:49:29.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2022-01-14 18:50:22.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2022-01-18 19:11:41.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"}],"messages":[{"id":"c8b2f0ffac9b9d2b9bca0819a6284ef61f8f56a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-06 10:48:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"263a4d66f40cb880c8427f20362c4343bcef915d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 12:11:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ebacc340eff47d881db8374a6a32336 : TIMED_OUT in 1h 02m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f459fd7299fb4fafa3992f4608df2f4b : TIMED_OUT in 1h 02m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a8f72c015beac5d804d5ca72d7046f23bb76be01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 12:33:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3454c4924ff748589dbd74c8e786b442 : SUCCESS in 15m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb9f6be2304f493b85394b41ed9199ec : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/080c01119362408487c1721d3f836854 : SUCCESS in 18m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1bbbf245babb4d35b8c1b789820971e8 : SUCCESS in 15m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e39afdad87d46f58ce5a56c29abaec8 : SUCCESS in 17m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/0205123d40874276a192e1ce67060301 : SUCCESS in 1h 00m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed10b77c18964fe69e71c520a7472f8c : SUCCESS in 1h 39m 06s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/a85200bac6b0493bb7e7afc61971824d : SUCCESS in 30m 12s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/3510a8b657a64cfab6f112669c0b9108 : SUCCESS in 47m 13s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d95f8d51e5aa45e5bc70b91fdc31bb61 : SUCCESS in 33m 42s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/b5164a1f5a8642b3965471240987b925 : SUCCESS in 31m 52s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c3cc45ba79304833b5860b64e4c81b4f : SUCCESS in 16m 28s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/f11f519db6a34255b1fd168341eb6cf7 : SUCCESS in 37m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c341193378047c480d6701a882dff8f : SUCCESS in 52m 45s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/953b6405842d4aa2996fbdd59d7b62c0 : SUCCESS in 41m 19s","accounts_in_message":[],"_revision_number":1},{"id":"27b90dff142b70f770bb7d34e5ca10406d0c9938","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-01-14 18:49:18.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2d190b2841fbc035a8b5eac11bd6452b4e63e935","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-01-14 18:49:29.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"19b7b814cf89ccb2b639002df2afc9bd62fd04f2","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-01-14 18:50:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e5280f67485cf7914fb910064dbd762c9c302179","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-14 19:14:29.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8d0456e11a03b017b0c1e77181fe5d0bf719518e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 20:59:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4d5b651ff5749a0b239bd420c3f4d44 : SUCCESS in 18m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ba49f7c000642e1b4bf04f8bbd3b561 : SUCCESS in 8m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffd559adb2314243a990fa96695f2666 : SUCCESS in 16m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93f242283a884b2ca91d12c28654856d : SUCCESS in 14m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52a8e79139984550a5cfc39034a1709b : SUCCESS in 15m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/73258ae37719431d8de4b6a98c360c50 : SUCCESS in 1h 10m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/16ef841762854a3e9c147d90e17792bb : SUCCESS in 1h 40m 41s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/93b09efd773b4435b63291d5b00da545 : SUCCESS in 41m 30s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/569ef4271e8b4cd596b45c24d832eec9 : SUCCESS in 49m 11s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a890808531244b68f46ceb8359b6e23 : SUCCESS in 33m 11s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/b1dd1e80370641b086e64af094a9e0f8 : SUCCESS in 37m 37s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/2370e23ccb6b4d1fb7e59de1143da1a2 : SUCCESS in 21m 41s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/a62097e6ca2446189b43b5e35b2dd77a : SUCCESS in 30m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a41e0713f533407280541ea0f7f38d21 : SUCCESS in 55m 57s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/d8f13a3ba7b248cbb9778e59e1f34975 : SUCCESS in 43m 58s","accounts_in_message":[],"_revision_number":2},{"id":"9edd08f4857ee3529645fe6ac95bc2aec24258eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 21:34:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/418442d872fb49758b1526f807a1a17c : TIMED_OUT in 1h 01m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2b00476490514f779962be013b36e6d8 : TIMED_OUT in 1h 01m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"838a35163171c4c06148fd4b3c706a5ad0f93b50","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-01-18 19:11:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"45c7fe025185fd2a0ac551eb8a28363ab4216149","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 19:13:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d268743f4541799c4d2da0d154bea9ab5b5313f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 20:48:25.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"0d5105b17b18cb4fc59de1df5f404da5dca86446","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-19 18:51:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"595ea8f7a62a64976455d9815390d8cecbae5b62","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-01-19 20:41:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"56f8ae9080723284ec41989ede71a7c8022023b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 20:43:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8e8d0c4f8a342e7871ee9b50eee0b73 : SUCCESS in 16m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc7c01db29fd46dca261da54d63d2647 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7543bed89d394bdd88aad6d3d227604c : SUCCESS in 13m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3eda2e3d7442479393e1697cd212e532 : SUCCESS in 14m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78dd6feaf72344a9b04e7942204d8f46 : SUCCESS in 14m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/7e7b86935a5d44a1860e8aa9c132ec5a : SUCCESS in 59m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ca90973ff4354d0aa63ebc3d9d64797c : POST_FAILURE in 1h 24m 45s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/c6cfdaf0d4f2450c83a56fd6c485b901 : SUCCESS in 28m 56s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/cb7d2e1c4fba4c42908c3745dd76eecb : SUCCESS in 37m 17s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6699730221df44f7a23fac1b7f7fca5d : SUCCESS in 27m 04s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/19086590b9444b9e9762c946c0065344 : SUCCESS in 27m 38s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/15157cfb1a944467bd670d11cadbbebf : SUCCESS in 14m 45s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/4baff698f4c94b71965321c094ab99b4 : SUCCESS in 34m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86350290f649405f96e6cf47f73dabda : SUCCESS in 46m 24s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/e70bbc471f7b483ca638d0c79ae38710 : SUCCESS in 40m 11s","accounts_in_message":[],"_revision_number":2},{"id":"8a886dd1cfc9a04f85409c518947a97595325ac3","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-01-19 20:45:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a79907a7507204243ede6e8b226db1121b05a277","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 21:20:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/abb903e4e6a34ab595458e3b337a09ec : TIMED_OUT in 1h 01m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/190fb0f289444e7a893448c908114c89 : TIMED_OUT in 1h 01m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cbdd8892f9ba7a2307c6edc538c5dc7a05226dea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 23:17:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c77181688be4913b11f49c4a94dde6d : SUCCESS in 27m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8471e94cbb23476cad733b8c8b288b69 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c81d096f12304d5582778c66f37f147f : SUCCESS in 25m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8791635746314103a15710a46d4023ef : SUCCESS in 25m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/336d6a9dd2324601acd52b5046954e4b : SUCCESS in 13m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/756fb77d64254eba8707fae03deb17e9 : SUCCESS in 1h 06m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/814656beae7144e29729316dd2a59bb6 : SUCCESS in 1h 43m 53s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/39e0ee789c6341b5b361232065cc1ccc : SUCCESS in 28m 16s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/7a4dbb665570445ba95fa5175d56a6aa : SUCCESS in 33m 57s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/620a28bad2144471b0577037234fadeb : SUCCESS in 31m 53s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/bb6059d6579749778a7d4097c7754b0f : SUCCESS in 31m 57s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/b976bbb52a3a47808909a536c4b70c61 : SUCCESS in 12m 54s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/f5be15ffc092439799236764fedf436a : SUCCESS in 32m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e48665724e8e448c9346c62725b443f0 : SUCCESS in 53m 01s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/ac1d4ed0282e4ab5bc2b5b95bd5fc1f4 : SUCCESS in 40m 53s","accounts_in_message":[],"_revision_number":2},{"id":"469dce27835efc6a5987eb98555dbd9c63b40427","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 23:49:17.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"df9f99a7eb49fc9d209265793d0728745aecb97f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-20 09:37:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5dd0cc5bce6f2152ec474e60529f715196e44489","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 10:46:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f49316d8d924a2a8c82478a84be6ba7 : TIMED_OUT in 1h 01m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d22c8f14736c438d9332127f9dd17c81 : TIMED_OUT in 1h 01m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a8809256cc2b06e5ba5fbaabcc316f28593c48ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 11:07:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57f98848f2654a5e8d0aab66fcfc7a44 : SUCCESS in 19m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/019ee1e6da124dd4898c6e1045fd6a98 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e73618798f44251ab985645acc335b8 : SUCCESS in 16m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d00f2ae163454695a3f2bf44f78cee97 : SUCCESS in 17m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2830be22d1d74979b54ae5b09dcfe274 : SUCCESS in 13m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/aac16905dd8546aa9c7717bfd62e853d : SUCCESS in 55m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec1ce82866034d8383fdda34ccfb3e0b : SUCCESS in 1h 16m 35s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/1d8db1c19c2d473f93b2eadb3dca33f1 : SUCCESS in 25m 53s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/fd11d9271fe648f5aaa7db9772259928 : RETRY_LIMIT in 4m 30s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b37eb11335d04ae28cf153de545471d2 : SUCCESS in 26m 10s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/1f7c2ce7bc004b86862cbac73cb6e011 : SUCCESS in 29m 26s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c6291a5ac3d64537bee7da6a4aa2035d : SUCCESS in 11m 36s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/98d14ae0073a46aeb3a3bb8ed84e3447 : SUCCESS in 33m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf562e29210449b886ebc426c3401112 : SUCCESS in 46m 35s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/4f083be712f74e6db4d490c513ae42cc : SUCCESS in 38m 24s","accounts_in_message":[],"_revision_number":2},{"id":"40f2d3c792d4b9a2fbd5d75fb553cba57a3e0c3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 11:08:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"69791d75dd1fb85b389e85a74adddfb5e2d55684","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 12:32:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37b52ed46b48443b98a95d7235e2f1b6 : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d14ae769d6de4933866a87565984c1b7 : SUCCESS in 15m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/759a01affc74450aa15a0fa23647c1bc : SUCCESS in 17m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab34721d3bd34d9ea0b5193aa4219098 : SUCCESS in 14m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/71b7b1811306463a9c6340f01baf994d : SUCCESS in 1h 01m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b964ce9569ee401d8fddbded3e32d3bc : SUCCESS in 1h 22m 26s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/2a727842dcab4c3390bc07616794cf88 : SUCCESS in 29m 47s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/753d082d950e423788515493bef078bc : SUCCESS in 30m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a62c7f324a94f6a975a079f69194552 : SUCCESS in 48m 34s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/2693f4ec19f9425dba0a6e5bbef86bd7 : SUCCESS in 46m 29s","accounts_in_message":[],"_revision_number":2},{"id":"c21ed9ad88aeb141f6e6f3a2da0819f80107be13","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 12:32:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8d73ea8de18241a890419e7ca6b9ff10b53ff9da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 12:35:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e19600add909435a9e0dae7712b29cc0 : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7e779ffec96a9b6eccb2f196c0f5cb070258f7b0","revisions":{"b7502adbe2bb9b5bfd8b0b58254b28434753319f":{"kind":"REWORK","_number":1,"created":"2022-01-06 10:48:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/63/823663/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/63/823663/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/63/823663/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/63/823663/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/63/823663/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/63/823663/1"}}},"commit":{"parents":[{"commit":"7a146850e0dec437a15b74afebeaec1bab88e8cc","subject":"sql: Squash mitaka migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7a146850e0dec437a15b74afebeaec1bab88e8cc"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-12-23 13:01:54.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-05 18:29:43.000000000","tz":0},"subject":"sql: Squash newton migration (part 1)","message":"sql: Squash newton migration (part 1)\n\nMake the following changes to the new \"initial\" migration.\n\n- Add \u0027nonlocal_user\u0027 table (103)\n- Add \u0027created_at\u0027, \u0027updated_at\u0027 columns to \u0027password\u0027 table (105)\n- Modify \u0027password\u0027 column of \u0027password\u0027 table to be nullable (106)\n- Add \u0027created_at\u0027 and \u0027last_active_at\u0027 columns to \u0027user\u0027 table (107)\n- Add \u0027failed_auth_count\u0027 and \u0027failed_auth_at\u0027 columns to \u0027local_user\u0027\n  table (108)\n- Add \u0027self_service\u0027 column to \u0027password\u0027 table (109)\n\nNote that 101 and 104 are not present since these were fixes for 088\nand 091 respectively, and they have no effect with newer versions of\nthe latter.\n\nSchemas can be compared using the techniques discussed at [1]. A sample\ncommand for SQLite:\n\n  # with change before this\n  python manage.py version_control \\\n    --database \u0027sqlite:///before.db\u0027 --repository . --version 097\n  python manage.py upgrade \\\n    --database \u0027sqlite:///before.db\u0027 --repository . --version 109\n\n  # with change after this\n  python manage.py version_control \\\n    --database \u0027sqlite:///before.db\u0027 --repository . --version 108\n  python manage.py upgrade \\\n    --database \u0027sqlite:///before.db\u0027 --repository . --version 109\n\nThis marks the end of migration squashing in for the legacy \u0027migrate_repo\u0027\nrepo. However, it\u0027s labelled part 1 because there are other newton-era\nmigrations in the other repositories. Future changes will focus on these\nother repos.\n\nChange-Id: Ib6ecae7a2e6fa3780106fa6a4b50d47aab922b43\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b7502adbe2bb9b5bfd8b0b58254b28434753319f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b7502adbe2bb9b5bfd8b0b58254b28434753319f"}]},"branch":"refs/heads/master"},"7e779ffec96a9b6eccb2f196c0f5cb070258f7b0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-01-14 19:14:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/63/823663/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/63/823663/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/63/823663/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/63/823663/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/63/823663/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/63/823663/2"}}},"commit":{"parents":[{"commit":"7a146850e0dec437a15b74afebeaec1bab88e8cc","subject":"sql: Squash mitaka migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7a146850e0dec437a15b74afebeaec1bab88e8cc"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-12-23 13:01:54.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-07 17:17:09.000000000","tz":0},"subject":"sql: Squash newton migrations (part 1)","message":"sql: Squash newton migrations (part 1)\n\nMake the following changes to the new \"initial\" migration.\n\n- Add \u0027nonlocal_user\u0027 table (103)\n- Add \u0027created_at\u0027, \u0027updated_at\u0027 columns to \u0027password\u0027 table (105)\n- Modify \u0027password\u0027 column of \u0027password\u0027 table to be nullable (106)\n- Add \u0027created_at\u0027 and \u0027last_active_at\u0027 columns to \u0027user\u0027 table (107)\n- Add \u0027failed_auth_count\u0027 and \u0027failed_auth_at\u0027 columns to \u0027local_user\u0027\n  table (108)\n- Add \u0027self_service\u0027 column to \u0027password\u0027 table (109)\n\nNote that 101 and 104 are not present since these were fixes for 088\nand 091 respectively, and they have no effect with newer versions of\nthe latter.\n\nSchemas can be compared using the techniques discussed at [1]. A sample\ncommand for SQLite:\n\n  # with change before this\n  python manage.py version_control \\\n    --database \u0027sqlite:///before.db\u0027 --repository . --version 097\n  python manage.py upgrade \\\n    --database \u0027sqlite:///before.db\u0027 --repository . --version 109\n\n  # with change after this\n  python manage.py version_control \\\n    --database \u0027sqlite:///before.db\u0027 --repository . --version 108\n  python manage.py upgrade \\\n    --database \u0027sqlite:///before.db\u0027 --repository . --version 109\n\nThis marks the end of migration squashing in for the legacy \u0027migrate_repo\u0027\nrepo. However, it\u0027s labelled part 1 because there are other newton-era\nmigrations in the other repositories. Future changes will focus on these\nother repos.\n\nChange-Id: Ib6ecae7a2e6fa3780106fa6a4b50d47aab922b43\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7e779ffec96a9b6eccb2f196c0f5cb070258f7b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7e779ffec96a9b6eccb2f196c0f5cb070258f7b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
