)]}'
{"id":"openstack%2Fkeystone~823666","triplet_id":"openstack%2Fkeystone~master~I59882d88fe593ec1ae37415b2157584f7f3c85f8","project":"openstack/keystone","branch":"master","topic":"bp/remove-sqlalchemy-migrate","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-02-04 22:41:29.000000000","reason":"Change was submitted"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2022-02-04 16:27:46.000000000","reason":"David Wilde replied on the change"}},"hashtags":[],"change_id":"I59882d88fe593ec1ae37415b2157584f7f3c85f8","subject":"sql: Remove legacy \u0027migrate_repo\u0027 migration repo","status":"MERGED","created":"2022-01-06 10:48:49.000000000","updated":"2022-02-04 22:44:06.000000000","submitted":"2022-02-04 22:41:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"823666-bp/remove-sqlalchemy-migrate","meta_rev_id":"b6fa55ae0f8da8de984e30e168286013aae1b736","_number":823666,"virtual_id_number":823666,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-04 22:41:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2022-02-04 16:30:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2022-02-04 16:27:46.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-19 20:42:06.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2022-02-04 16:30:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-06 12:11:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-06 12:38:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-14 18:55:53.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2022-01-19 20:42:06.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2022-02-04 16:30:35.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"}],"messages":[{"id":"1c3985d3f4eb65495bd4a43257467dda7207d191","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-06 10:48:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5b104fd22b7db21216cf188eea7c3ab524441cb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-06 11:07:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f0c3c79c3620de72dca354a08fe4312c5db12cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 12:11:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bb4883ee39854459aa364e5d086a1bcf : SUCCESS in 44m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c9e104a7392b40e8aa6cb29edfe13cb7 : TIMED_OUT in 1h 02m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"65b4433438844ce27dd9da1194e5308349e43573","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 12:38:42.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/019cfa41b8d545b98f440c8f3ef1cfba : SUCCESS in 18m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/584f6f24a4ed47b686c386320dc2ae36 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/483e7623e99a43fb89d179510dd767ac : SUCCESS in 22m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39fd366bdeb84e2aabfcde469462e8c4 : SUCCESS in 22m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28f6ed80672945139e3e07be735f6c92 : FAILURE in 14m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/a91b3769dc774c8db5f09331707194ae : SUCCESS in 1h 09m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/873f751565bd4aa59c464bd2bbac1f01 : SUCCESS in 1h 33m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e8907ba3a6146009ed02d439fc72469 : SUCCESS in 9m 31s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/b9affd29e80845d695c4d69d3fd182e7 : SUCCESS in 29m 25s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/167d2dc758c74e5eb0d5109f68a74b93 : SUCCESS in 36m 29s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e185d64bc1f4f0ab8ded1bd2ab62d5b : SUCCESS in 39m 55s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/2fd3ed94106a4409be1179225da05d18 : SUCCESS in 38m 21s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d09e73666ccc4452ac83e312c176cf8a : SUCCESS in 18m 39s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/09c69acc5de24e6ca9160d781b83f669 : SUCCESS in 38m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20aeea3414fd42a2b110b85fa417c348 : SUCCESS in 52m 52s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/949e226a376744a09a89783f773d374a : SUCCESS in 42m 23s","accounts_in_message":[],"_revision_number":1},{"id":"e7c07aff2679a12f8b3f2771f6db05151d7d0605","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-01-14 18:55:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9610308c15b2b7e8e2eca7132b3b31838b172ccc","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-01-14 18:56:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d27604a53007b26eefd172b8693728d13fe6f903","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-14 19:14:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bb49a77dfb1846a660ebd1f017509b9481394379","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-14 19:21:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"41cd044c2dddf83160f3d4c8d01a88888eec747e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-14 19:23:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f546ccf54919a91deed641879c404c5779a105d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 20:28:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/89ac8ee84ca949d3927531ea586a1412 : TIMED_OUT in 1h 01m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f53a40d8dd242a6a3a4c7b59a1c8e71 : TIMED_OUT in 1h 02m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"29b83e348283c7d356897d085ae8d3e2ac953688","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 20:47:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07010efc0de941dabde6c73d30880274 : SUCCESS in 15m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de404b222fea425998e1f92156f42cbd : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30fec5d9ef7e4b82a4bd4ef2a31ccbe8 : SUCCESS in 11m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aec8c019bb3e454ca42a47b18abdaa3d : SUCCESS in 13m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/210c435206574b6b8f72c591b12292c8 : SUCCESS in 12m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/a8d79a30802a45d8b437c3429fe3a140 : SUCCESS in 49m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/542b157ece774b0a89196ebf9b24e69f : SUCCESS in 1h 19m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78f7a0bc85944100aa9b296ed88eca2c : SUCCESS in 9m 12s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/54238d08666744ad9ba1f7445661875e : SUCCESS in 27m 20s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/60ebce537a8242d7aa432380b0021eca : SUCCESS in 40m 56s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34c187623c184b3fb02ffb34d3757d1c : SUCCESS in 24m 30s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/3a70e2da14774b0a8811a98e266204be : SUCCESS in 24m 26s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/537886dac2de408d9b85a68405a3f74b : SUCCESS in 13m 53s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ee0ce7041f3845a2bb4a01b854dc1ea3 : SUCCESS in 28m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aea53e10a8144042b688cf48a8ea89c9 : SUCCESS in 56m 21s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/6051638b60a7463fa8cec25d33b426ee : SUCCESS in 39m 20s","accounts_in_message":[],"_revision_number":2},{"id":"d63111c243825253983b3acce6287ee580b6753c","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-01-19 20:42:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"10df2beb1f0905baa1b4eb07411a62681c6a5c6b","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-02-04 16:27:46.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d5bbcab215b6c3fed2f6e7b30534263f95337559","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-02-04 16:30:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9dfc83d617e28764ed5f72b2c0a96b4d96d642b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 16:32:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b739eb7b4159698b8dc31d7e5736192711cdb568","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 22:41:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdc70afed4b94c309922f254aa9d3db9 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d72c90a6a1f548d78d7dae708bd496e2 : SUCCESS in 27m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a5e36ce708244779c023ba5f082fa9b : SUCCESS in 26m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52c80bae2d3e46d7af24880b582b6d16 : SUCCESS in 14m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/827f4b7985b7477c88d2bff965a55e8d : SUCCESS in 43m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7588958db6ee4aab8fb46bac6195f9d5 : SUCCESS in 1h 18m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/079ce3a9bf7b4b86b6b851efb574bf66 : SUCCESS in 7m 54s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f5a90fb5ebd84e9bb9dc7ea4e23cf903 : SUCCESS in 27m 15s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/c3389c065f0346d9adefd9a537443ca9 : SUCCESS in 28m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ebff174e40d343d182bc34917f8ff07c : SUCCESS in 43m 06s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/bfbb8851adbb464386e04946b4b4f123 : SUCCESS in 40m 31s","accounts_in_message":[],"_revision_number":2},{"id":"49e17c26a12fb9a72ace0bfa368a77d06311e8f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 22:41:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b6fa55ae0f8da8de984e30e168286013aae1b736","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 22:44:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2acc8ad3522a459e8f9779e4d0e5b131 : SUCCESS in 1m 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e10b4022cdd41f98f4d2d165d2afb0c : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"503421d3d492cad1bd110ec72870f05d80fe021c","revisions":{"d18c88c84be3431a78ae7fef2be9be6309f9f930":{"kind":"REWORK","_number":1,"created":"2022-01-06 10:48:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/823666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/66/823666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/823666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/823666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/823666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/66/823666/1"}}},"commit":{"parents":[{"commit":"42ec710e8d49c179ea20f340478a66b34c0cc3aa","subject":"sql: Fold unique constraints into table definitions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/42ec710e8d49c179ea20f340478a66b34c0cc3aa"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-12-23 19:09:37.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-05 18:29:43.000000000","tz":0},"subject":"sql: Remove legacy \u0027migrate_repo\u0027 migration repo","message":"sql: Remove legacy \u0027migrate_repo\u0027 migration repo\n\nThis is now folded into the initial migration of the \u0027expand_repo\u0027\nrepository. Previously, this was a dummy migration. We simply move\nthings across and remove any code that was trying to work with the older\nrepo since it\u0027s no longer necessary.\n\nA release note is added, even though it\u0027s not really necessary since\nnothing will change for users. It\u0027s more of a heads up that things are\nafoot.\n\nChange-Id: I59882d88fe593ec1ae37415b2157584f7f3c85f8\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d18c88c84be3431a78ae7fef2be9be6309f9f930"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d18c88c84be3431a78ae7fef2be9be6309f9f930"}]},"branch":"refs/heads/master"},"503421d3d492cad1bd110ec72870f05d80fe021c":{"kind":"REWORK","_number":2,"created":"2022-01-14 19:14:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/823666/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/66/823666/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/823666/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/823666/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/66/823666/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/66/823666/2"}}},"commit":{"parents":[{"commit":"7695d81a412ca340fa9773156000f04c0aaa592e","subject":"sql: Fold unique constraints into table definitions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7695d81a412ca340fa9773156000f04c0aaa592e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-12-23 19:09:37.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-10 18:31:09.000000000","tz":0},"subject":"sql: Remove legacy \u0027migrate_repo\u0027 migration repo","message":"sql: Remove legacy \u0027migrate_repo\u0027 migration repo\n\nThis is now folded into the initial migration of the \u0027expand_repo\u0027\nrepository. Previously, this was a dummy migration. We simply move\nthings across and remove any code that was trying to work with the older\nrepo since it\u0027s no longer necessary.\n\nA release note is added, even though it\u0027s not really necessary since\nnothing will change for users. It\u0027s more of a heads up that things are\nafoot.\n\nChange-Id: I59882d88fe593ec1ae37415b2157584f7f3c85f8\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/503421d3d492cad1bd110ec72870f05d80fe021c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/503421d3d492cad1bd110ec72870f05d80fe021c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
