)]}'
{"id":"openstack%2Fkeystone~824781","triplet_id":"openstack%2Fkeystone~master~Ic499a996e1b9c6f0a9d885b4a9e655008002d49f","project":"openstack/keystone","branch":"master","topic":"bp/remove-sqlalchemy-migrate","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-02-08 10:48:14.000000000","reason":"stephenfin replied on the change"}},"hashtags":[],"change_id":"Ic499a996e1b9c6f0a9d885b4a9e655008002d49f","subject":"sql: Squash queens migrations","status":"MERGED","created":"2022-01-14 19:14:29.000000000","updated":"2022-02-08 14:17:05.000000000","submitted":"2022-02-08 13:45:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824781-bp/remove-sqlalchemy-migrate","meta_rev_id":"919b5999d817170cbf8429eeb1d8339f999b26c6","_number":824781,"virtual_id_number":824781,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-08 13:45:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2022-02-04 16:44:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2022-02-04 16:44:56.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2022-02-04 16:45:04.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2022-02-04 16:48:17.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2022-02-04 16:48:17.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-14 21:06:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-14 21:52:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-04 16:44:38.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2022-02-04 16:44:56.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2022-02-04 16:45:04.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2022-02-04 16:48:17.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"}],"messages":[{"id":"d40c9fdb5228321c0af18c3c8e49173311994109","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-14 19:14:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f0b5775782b218e9485d2a2800907acbcee02f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 21:06:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/53c13ad8825d4abb8371cf0762611a4b : SUCCESS in 50m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b118418ca43148e2b417f5800669ef4e : TIMED_OUT in 1h 02m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5ab877505363079b510e2358e7b1c05f62e14dfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 21:52:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/047b4aa0a77c44a19d29e029857983d1 : SUCCESS in 17m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d291ad57c5a841cda9e042eb069f9f82 : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fd5e2a11933438aad2aba0ccd3296f2 : SUCCESS in 13m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0be0e6fd171443469e36f673ddfa243f : SUCCESS in 19m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f87571b876d744ea8767e4e6c704577e : SUCCESS in 16m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/6428a110adc7469081118325787aefdd : SUCCESS in 1h 08m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e8475a89bbf4b4bbbe3a2a4746280eb : FAILURE in 25m 05s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/fd46235608b64daa905c7587db66c4b6 : FAILURE in 21m 17s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/898004144c9b47f1a66bcdd69684eaf9 : FAILURE in 36m 11s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/829742780bfd4f8bb82536cea8a63ff0 : FAILURE in 25m 06s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/2fd86e6bc5de4e4f8d64de16cb8c410b : FAILURE in 25m 31s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/ff2647fb0de7469eb7f02b245bce43ee : POST_FAILURE in 13m 50s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/9b1a83ae8ef74b02b16295517c0bfe88 : FAILURE in 28m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c90161846d94ef08b395b9fbb1eff1e : FAILURE in 25m 59s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/e28884f4716b4f0cb4cfeccb2b36e837 : FAILURE in 20m 49s","accounts_in_message":[],"_revision_number":1},{"id":"0c90409e3b225f65ca43e0080d584ab693266343","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-19 18:50:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"53fb7a15476e80657432f194e5431ab4c92871ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 20:29:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/348d5ba0f30d47498586997404561fc1 : TIMED_OUT in 1h 03m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9a921febe41f40b5b4b4b6207c02a4fb : TIMED_OUT in 1h 02m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3b8fa655a81b0dd5f8bc5e3689579ece48baa013","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 20:55:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2214e92de0ad44d48cd7e405db09fc5b : SUCCESS in 17m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cd1d83819074eccad19d9e00d9f0108 : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09dc8420aff04790b64815bdc5bce15a : SUCCESS in 11m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f95570ae75d450994d718144afe274f : SUCCESS in 13m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84224fca848b4bdebee67f38c68bcdea : SUCCESS in 14m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/8ad62266251b43a3ab96113d1a099b2b : SUCCESS in 56m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ebd82caa905409fb2d1882e3831b939 : SUCCESS in 1h 49m 41s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/73f4ddbbbd284e1f9637d434661522f5 : SUCCESS in 42m 44s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/c9e82f7a0bb84e0a8aa3aff696fd412e : SUCCESS in 37m 14s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddf59e176e7c480abf5ccc215b2ab9a0 : SUCCESS in 39m 57s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/df7337e79a054625b1a669747f54364d : SUCCESS in 24m 45s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/148caf2a7aeb4eddae69eab7740fd73b : SUCCESS in 20m 28s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/a45880717a9640d2a06e5d7852135575 : SUCCESS in 52m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fbb4325199d746c9ae58349e5ed09e4e : SUCCESS in 47m 36s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/61ae24cb614d4021821672e0f92795b6 : SUCCESS in 49m 23s","accounts_in_message":[],"_revision_number":2},{"id":"48c1bc385dee3dbaccaaf8a6d9af210cc42f1957","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-02-04 16:44:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f1d455aac2ffd7c0fbf3b951fb07ced38b195947","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-02-04 16:44:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d5aad28d73ddb2c66c1b8c39f8954d0915756649","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-02-04 16:45:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2ab210ac65244b646b82c5e5b7a581c4ed09641b","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-02-04 16:48:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9e52bd056df8523e81d43e8eb919b56fe751d589","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 20:06:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"738a0de679e34afedc4388d7f3ff7b39c050724d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 21:29:34.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"8c4eb706c0145cd9d899615a90426ed75a981b2a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-08 10:48:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a8705f5956db4ec612d8eee72f87778379c4a5ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 12:22:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eec73c79718a4d86be1a32dcc82ec383 : SUCCESS in 16m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a6c5328009746aaa3cd1b330d9cbc9b : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfb390eeef96413b978606174cefe350 : SUCCESS in 18m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9c2655ed23e4547a2b9532a5c4b9bd3 : SUCCESS in 14m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/912f270fc1e74f3a981fdf7da3a59a3f : SUCCESS in 14m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/311493713c614084876f1bd7bc08bfac : SUCCESS in 56m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4cdf0047b49e48ef81681ebbf4137bf2 : SUCCESS in 1h 29m 56s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/b0e427a8aed74c0dbf1faf0642866d9f : SUCCESS in 29m 11s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/522ad30678134ae1a2dc918997be11de : SUCCESS in 36m 01s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/196f00097ddf4f7dae6aefc32860ed91 : SUCCESS in 28m 23s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/963004a099f54b038820077ae3835d94 : SUCCESS in 28m 30s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d14d2ff28cc14851a82340d2df05253a : SUCCESS in 10m 44s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/569ce911a2df4d15b269aa6a02e1eae3 : SUCCESS in 35m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f01bbf7a33954e12ada006703c1fda51 : SUCCESS in 40m 54s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/4ee2f701d62f4969ba48a1371a32cd9e : SUCCESS in 40m 44s","accounts_in_message":[],"_revision_number":2},{"id":"fa9032e48836a09d2ae9610fa2bf03506e353460","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 12:23:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3ad08114df32f383c603679e7dacc905a306750a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 13:45:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3472b1b78934effbb0ac7270e3ab306 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25b628147ab6472cb61abf8e0ed71c14 : SUCCESS in 10m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b046da401a944f1b8e1ab48cc389544c : SUCCESS in 12m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d00d2540b3d47b1a47231ec55005a8a : SUCCESS in 11m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/aafb5a40799042e3a91275367ec71016 : SUCCESS in 1h 00m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/10336285fdb14114b182d8d97443739e : SUCCESS in 1h 19m 02s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/40cd93adf36f4b64a69a765b88d24406 : SUCCESS in 36m 03s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/4c55755636474675aabb1d48f27e0d4b : SUCCESS in 34m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ede4b87c1d8447c6aa89bf160e16da0d : SUCCESS in 46m 34s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/8ca3bf3ab02e4a418934ddbbacf517c4 : SUCCESS in 39m 05s","accounts_in_message":[],"_revision_number":2},{"id":"57a34786dec69c59409bee633f683c88de44f942","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 13:45:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4f1edd95f8ba10b85e32449e6d0c0f829feefc9b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 13:55:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94f5079eb52a42e795a16b81c644058e : SUCCESS in 2m 04s","accounts_in_message":[],"_revision_number":2},{"id":"919b5999d817170cbf8429eeb1d8339f999b26c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 14:17:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b069f49e4604218a89f389b3b1bcaf6 : TIMED_OUT in 1h 01m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b7b692a502bd4543a95c21b7ec101df7 : TIMED_OUT in 1h 04m 35s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6058ae332a9ef49bd8523a186479773592119a0e","revisions":{"105b2584aaa15a8941fb8978ca7fef990f3b6cbc":{"kind":"REWORK","_number":1,"created":"2022-01-14 19:14:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/824781/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/81/824781/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/824781/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/824781/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/824781/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/81/824781/1"}}},"commit":{"parents":[{"commit":"09ec9d5392e70b78caecc2ef3f791b0aed04986f","subject":"sql: Squash pike migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/09ec9d5392e70b78caecc2ef3f791b0aed04986f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-08 16:55:52.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-11 18:10:00.000000000","tz":0},"subject":"sql: Squash queens migrations","message":"sql: Squash queens migrations\n\nMake the following changes to the new \"initial\" migrations.\n\n- Add \u0027project_tag\u0027 table (030)\n- Add \u0027system_assignment\u0027 table (031)\n- Add \u0027expires_at_int\u0027 column to \u0027trust\u0027 table and expand the unique\n  constraint covering \u0027trustor_user_id\u0027, \u0027trustee_user_id\u0027,\n  \u0027project_id\u0027, \u0027impersonation\u0027, and \u0027expires_at\u0027 to also cover the\n  \u0027expires_at_int\u0027 column (032)\n- Add \u0027limit\u0027 and \u0027registered_limit\u0027 tables (033)\n- Add \u0027application_credential\u0027 and \u0027application_credential_role\u0027 tables\n  (034)\n- Add \u0027system\u0027 column to \u0027application_credential\u0027 table and make\n  \u0027project_id\u0027 column nullable (035)\n- Rename \u0027allow_application_credential_creation\u0027 column of\n  \u0027application_credential\u0027 table to \u0027unrestricted\u0027 (036)\n- Remove foreign key constraints from \u0027service_id\u0027 and \u0027region_id\u0027\n  columns of \u0027registered_limit\u0027 table (037)\n\nChange-Id: Ic499a996e1b9c6f0a9d885b4a9e655008002d49f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/105b2584aaa15a8941fb8978ca7fef990f3b6cbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/105b2584aaa15a8941fb8978ca7fef990f3b6cbc"}]},"branch":"refs/heads/master"},"6058ae332a9ef49bd8523a186479773592119a0e":{"kind":"REWORK","_number":2,"created":"2022-01-19 18:50:03.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/824781/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/81/824781/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/824781/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/824781/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/824781/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/81/824781/2"}}},"commit":{"parents":[{"commit":"677303732f2fd310bba65c62b905cd8f510ff6e0","subject":"sql: Squash pike migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/677303732f2fd310bba65c62b905cd8f510ff6e0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-08 16:55:52.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-19 16:58:16.000000000","tz":0},"subject":"sql: Squash queens migrations","message":"sql: Squash queens migrations\n\nMake the following changes to the new \"initial\" migrations.\n\n- Add \u0027project_tag\u0027 table (030)\n- Add \u0027system_assignment\u0027 table (031)\n- Add \u0027expires_at_int\u0027 column to \u0027trust\u0027 table and expand the unique\n  constraint covering \u0027trustor_user_id\u0027, \u0027trustee_user_id\u0027,\n  \u0027project_id\u0027, \u0027impersonation\u0027, and \u0027expires_at\u0027 to also cover the\n  \u0027expires_at_int\u0027 column (032)\n- Add \u0027limit\u0027 and \u0027registered_limit\u0027 tables (033)\n- Add \u0027application_credential\u0027 and \u0027application_credential_role\u0027 tables\n  (034)\n- Add \u0027system\u0027 column to \u0027application_credential\u0027 table and make\n  \u0027project_id\u0027 column nullable (035)\n- Rename \u0027allow_application_credential_creation\u0027 column of\n  \u0027application_credential\u0027 table to \u0027unrestricted\u0027 (036)\n- Remove foreign key constraints from \u0027service_id\u0027 and \u0027region_id\u0027\n  columns of \u0027registered_limit\u0027 table (037)\n\nChange-Id: Ic499a996e1b9c6f0a9d885b4a9e655008002d49f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6058ae332a9ef49bd8523a186479773592119a0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6058ae332a9ef49bd8523a186479773592119a0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
