)]}'
{"id":"openstack%2Fkeystone~824782","triplet_id":"openstack%2Fkeystone~master~I6beb8c9c827546757f2d7673a917b98f8b917012","project":"openstack/keystone","branch":"master","topic":"bp/remove-sqlalchemy-migrate","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-02-08 10:48:20.000000000","reason":"stephenfin replied on the change"}},"hashtags":[],"change_id":"I6beb8c9c827546757f2d7673a917b98f8b917012","subject":"sql: Squash rocky migrations","status":"MERGED","created":"2022-01-14 19:14:29.000000000","updated":"2022-02-08 13:57:18.000000000","submitted":"2022-02-08 13:52:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824782-bp/remove-sqlalchemy-migrate","meta_rev_id":"be0830d1ebab6609e4abbc8bf7c34cf39ef4d31c","_number":824782,"virtual_id_number":824782,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-08 13:52:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2022-02-07 20:03:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2022-02-04 16:47:31.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2022-02-04 16:47:38.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2022-02-04 16:50:34.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2022-02-07 20:03:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-14 20:32:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-14 20:42:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-04 16:47:31.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2022-02-04 16:47:38.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2022-02-04 16:50:34.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2022-02-07 20:03:05.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"}],"messages":[{"id":"cdb3bd6a4c23d3976e2620487b3ad39df8d7a369","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-14 19:14:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1576f9c4cabb9e16d43d6e83bf5220663864df76","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 20:32:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0bbd93cc235c4e28848bd2a284c3181b : SUCCESS in 42m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c75e6ab1dff44922ad92df131799a221 : SUCCESS in 39m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26957a744dd2f88d19f1d31f75077cdb7122e944","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 20:42:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f59c0507a0324b8fbfbc9c247d663c12 : SUCCESS in 19m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea995188afd74875b581c2e6847a99c7 : SUCCESS in 8m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8ea6221aea34339ae0b30550d9f8e9e : SUCCESS in 16m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19c5ef837f8045f3975bab5c36765974 : SUCCESS in 14m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da3ee0d772fa45c5805dd55038ab0c8b : SUCCESS in 17m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/13db5a82735a414c98538c38acc4f73a : SUCCESS in 1h 03m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1af56b4079f54c36b44212133178a604 : FAILURE in 27m 40s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/9a11d746bc7e4afcaa766c7c04b81420 : FAILURE in 26m 12s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/4f193863e46345558c6ba476a73f2af4 : FAILURE in 26m 49s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7306bc7c70354413b49627c5bbabcef3 : FAILURE in 27m 02s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/ecc9b25e02ea4fe8ab06e78328d27fe4 : FAILURE in 25m 35s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/9f14346a69bf40a9bb35c745d3ffb15c : POST_FAILURE in 17m 47s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/3f868997806e4ed5bc5bbaaed084043e : FAILURE in 19m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51cb22ecfb2446e384397d1e30286e64 : FAILURE in 28m 01s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/c693b75157cb4ec2bf31103216f1ce61 : FAILURE in 18m 24s","accounts_in_message":[],"_revision_number":1},{"id":"48c549dcbbc5d7cd5f4247f8fa03b8b1d48fcbfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-19 18:50:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4a10f6813fa8357f7b4fa8c155520851542fcc37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 20:24:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8706b908eaf443b89aff797de0dc390f : TIMED_OUT in 1h 02m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/366a9708f8234fb6bb029eb1d4f99093 : TIMED_OUT in 1h 02m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"573180c9a193767f1243f7851340fa8a38f71df4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 20:31:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e0ee0946dbf47e983ebadfba642f4fb : SUCCESS in 16m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21b0bf9bc510439796c6a1a11ec242b7 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ade2427cb4994b9eb0f3442e21c6dcad : SUCCESS in 13m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/016153e6be1f47a3b4946dabd407f878 : SUCCESS in 19m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffdbad3647ca47e48da03c6f1d6dc2e3 : SUCCESS in 13m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/49a8b6a64e0c45f2bbc29988d2b25906 : SUCCESS in 1h 03m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6daf76b5c9464de78aab06edd91cd51e : SUCCESS in 1h 28m 19s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/581f85230754466b90be180836a5ef4f : SUCCESS in 35m 39s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/32febfa655084af4b5f12f8af023f64a : SUCCESS in 31m 35s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2006d169241d4b72930c0cf72539e361 : SUCCESS in 39m 26s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/19a2deb9bf9349a29e510fe169cdc624 : SUCCESS in 33m 58s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/171ac17002344d379d67f673e48b573f : SUCCESS in 12m 22s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/fbc50e3716884e88b717c7a174d8ba92 : SUCCESS in 43m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/091d6ae2a3dc414aab3b6fa87a3d5c51 : SUCCESS in 55m 07s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/054cbca8ceb743938a214437f5263009 : SUCCESS in 51m 03s","accounts_in_message":[],"_revision_number":2},{"id":"bb1d651e8a0be0f7a79a475a27f835c73bcb2a6c","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-02-04 16:47:31.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"27ff985eac7df00bf95505853feb6a091363db26","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-02-04 16:47:38.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a467cee5dc076e1caa94791f8cf3c9103f3cd4b7","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-02-04 16:50:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0dcd066727ba1f62d0ca9cbed6c8d93fd390d520","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-02-07 20:03:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"270f388f8f6af0442564dd86e6cf491632d14422","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 20:06:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2683472df4dcc3845c29ed0c3db7d9967d9d9872","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 21:29:35.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"fc7ed66e546044fee5df25f6db78b66c18459d13","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-08 10:48:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fc1b1370b6a79e5f21fff0e5fd590697d7dc1b14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 12:19:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38d18f12c48549c0851014454d82f123 : SUCCESS in 15m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a94cc7b7223149e0af5aa125f4c88469 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bb5f39a6bcb493e94c6915ee47a6ea4 : SUCCESS in 13m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45f3a52aac5541c98f99493f6532c73a : SUCCESS in 16m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a60412ea74a544df96f5eda2f3e36d81 : SUCCESS in 14m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/be28ba05e1f34b08899f34c4aa189179 : SUCCESS in 52m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9023c47fa4044c0896e086a3ab5596cd : SUCCESS in 1h 25m 42s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/ea6c4aedfe0747699c340f8a375d9166 : SUCCESS in 28m 55s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/b8683616c16542c29ff5d2c7e13deb87 : SUCCESS in 36m 14s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8b750460055492d925c6beba8c6af05 : SUCCESS in 31m 50s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/ad671fc4eb2c44fb8da98a1472add267 : SUCCESS in 31m 43s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/35a53c7168e7458db9586a0fb2536de4 : SUCCESS in 13m 21s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/296dab4307c8491da93c44a57bd09e40 : SUCCESS in 33m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b270cd29f58941dd9f2ba63c891df907 : SUCCESS in 49m 00s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/1356620fcbc4468f8cd24c13ddc05ab5 : SUCCESS in 38m 58s","accounts_in_message":[],"_revision_number":2},{"id":"1ee4d5323213ffa0abf1d7408005d8ff12eea2e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 12:23:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bea7b84ca2274c43172ba0a7687ce3719779f53f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 13:03:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/79bfcc3027bc45318cc508909d382c08 : SUCCESS in 58m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c41fe7efbf97413f861c84a5c3c3f66e : SUCCESS in 53m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"225764e8bbcfa9f53d21243b9769f216a59a0ccc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 13:52:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aac106253cec477fa25904ac29730c2d : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c963e10f752643818597f97a770d626b : SUCCESS in 10m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7fe435790c948f7b596680bfaa371ed : SUCCESS in 11m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58e5c64dbd354e5abf34e0251caa5988 : SUCCESS in 11m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/49f2aac8c19b4798ab0deb2a5aede562 : SUCCESS in 56m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0abd7726656245939c5fe2bddb5bcc73 : SUCCESS in 1h 25m 57s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/bd75e5f0d75c49d5995f973beaa4b562 : SUCCESS in 29m 50s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/b63231c876364b758954d9e484cb5d9e : SUCCESS in 25m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96dead623c294f82a03ebbded5f38fe9 : SUCCESS in 46m 47s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/e797da4837b04f34bc4f376f44ff1bae : SUCCESS in 40m 53s","accounts_in_message":[],"_revision_number":2},{"id":"bd0908aea16cd31e1041f936d32dfd932ed1d8d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 13:52:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"be0830d1ebab6609e4abbc8bf7c34cf39ef4d31c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 13:57:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/437b3857480b4b19aa4fff1848b165df : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"30ec0df2a3c656df686ceaca5f66d0067b8b007b","revisions":{"0ad8029778cc9294c1330e486c12dc09666fe4e3":{"kind":"REWORK","_number":1,"created":"2022-01-14 19:14:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/824782/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/82/824782/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/82/824782/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/82/824782/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/82/824782/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/82/824782/1"}}},"commit":{"parents":[{"commit":"105b2584aaa15a8941fb8978ca7fef990f3b6cbc","subject":"sql: Squash queens migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/105b2584aaa15a8941fb8978ca7fef990f3b6cbc"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-10 13:05:12.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-11 18:10:01.000000000","tz":0},"subject":"sql: Squash rocky migrations","message":"sql: Squash rocky migrations\n\nMake the following changes to the new \"initial\" migrations.\n\n- Add \u0027description\u0027 column to the \u0027limit\u0027 and \u0027registered_limit\u0027 tables\n  (045)\n- Add \u0027internal_id\u0027 column to the \u0027limit\u0027 and \u0027registered_limit\u0027 tables\n  and make \u0027id\u0027 column simply unique instead of the primary key (047)\n- Drop foreign key constraint on \u0027project_id\u0027 column of \u0027limit\u0027 table,\n  and drop unique constraints from both \u0027limit\u0027 and \u0027registered_limit\u0027\n  tables (047)\n- Add \u0027registered_limit_id\u0027 column to \u0027limit\u0027 table (048)\n\nChange-Id: I6beb8c9c827546757f2d7673a917b98f8b917012\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0ad8029778cc9294c1330e486c12dc09666fe4e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0ad8029778cc9294c1330e486c12dc09666fe4e3"}]},"branch":"refs/heads/master"},"30ec0df2a3c656df686ceaca5f66d0067b8b007b":{"kind":"REWORK","_number":2,"created":"2022-01-19 18:50:03.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/824782/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/82/824782/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/82/824782/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/82/824782/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/82/824782/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/82/824782/2"}}},"commit":{"parents":[{"commit":"6058ae332a9ef49bd8523a186479773592119a0e","subject":"sql: Squash queens migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6058ae332a9ef49bd8523a186479773592119a0e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-10 13:05:12.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-19 16:58:16.000000000","tz":0},"subject":"sql: Squash rocky migrations","message":"sql: Squash rocky migrations\n\nMake the following changes to the new \"initial\" migrations.\n\n- Add \u0027description\u0027 column to the \u0027limit\u0027 and \u0027registered_limit\u0027 tables\n  (045)\n- Add \u0027internal_id\u0027 column to the \u0027limit\u0027 and \u0027registered_limit\u0027 tables\n  and make \u0027id\u0027 column simply unique instead of the primary key (047)\n- Drop foreign key constraint on \u0027project_id\u0027 column of \u0027limit\u0027 table,\n  and drop unique constraints from both \u0027limit\u0027 and \u0027registered_limit\u0027\n  tables (047)\n- Add \u0027registered_limit_id\u0027 column to \u0027limit\u0027 table (048)\n\nChange-Id: I6beb8c9c827546757f2d7673a917b98f8b917012\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/30ec0df2a3c656df686ceaca5f66d0067b8b007b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/30ec0df2a3c656df686ceaca5f66d0067b8b007b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
