)]}'
{"id":"openstack%2Fkeystone~824785","triplet_id":"openstack%2Fkeystone~master~Ia1495cd4683d6631be2691e816734d01b03037a3","project":"openstack/keystone","branch":"master","topic":"bp/remove-sqlalchemy-migrate","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-02-08 15:57:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia1495cd4683d6631be2691e816734d01b03037a3","subject":"sql: Squash ussuri migrations","status":"MERGED","created":"2022-01-14 19:14:29.000000000","updated":"2022-02-08 16:18:00.000000000","submitted":"2022-02-08 15:57:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824785-bp/remove-sqlalchemy-migrate","meta_rev_id":"1553f3fcc227ae37cb21da0d476d6c1c515869ea","_number":824785,"virtual_id_number":824785,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-08 15:57:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2022-02-08 14:14:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2022-02-04 16:55:00.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2022-02-04 16:55:33.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-04 16:56:22.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2022-02-08 14:14:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-14 20:44:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-04 16:55:00.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2022-02-04 16:55:33.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2022-02-04 16:56:22.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2022-02-08 14:14:56.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"}],"messages":[{"id":"082b4e528a4ac90b481bcc5aef097dd88ffbd2e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-14 19:14:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67fbeaf9a19eb5f4f9984d5dadcc9c3e142a0f17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 20:44:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e74764b49ab4476ab17c3fdf3b69d464 : SUCCESS in 23m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e80547edd7104ce18d202786d1b0a3f3 : SUCCESS in 8m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45dee597cac448329d4d18a08495b178 : SUCCESS in 15m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90c99c72e2ae4e60b8868178029c6038 : SUCCESS in 15m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d106e17b30843aba6f5b57e4b43765d : SUCCESS in 13m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/78e5f8dddd9540e7ade9f567ed82fe80 : SUCCESS in 1h 03m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0fcd365e76184a4bbfe7eac8ba8c9cc2 : FAILURE in 24m 24s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/89272ee5e1464dafaece2e57e13958c0 : FAILURE in 25m 05s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/5f61ee1306274ba682d0c74bf6e878bc : FAILURE in 32m 46s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9fe99eaa4104be797293ce31259728d : FAILURE in 24m 42s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/fd78c174bcbb42d9974d32a7d4d390ba : FAILURE in 25m 43s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/4a3bfc6d50a64c0b99b92bfc18191a17 : POST_FAILURE in 18m 21s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/cccefbcaaf864ae89e70d8697f9e924c : FAILURE in 24m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0fc6023fac2540fcacc4240ddfabcaa3 : FAILURE in 24m 11s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/e712898ec9d842e3a9d2ba9995160d6a : FAILURE in 21m 16s","accounts_in_message":[],"_revision_number":1},{"id":"175391fbc051842257dd0702491940b6a7a0bc9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 21:00:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a9e16e2830e645f699c4200b7b851b96 : SUCCESS in 30m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c1976497371640f095bfd9389e4cd8ef : TIMED_OUT in 1h 02m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"31f989270f689e5b0a4acd9d01b34260e1ad8243","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-19 18:50:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"36b7f0706049905ebe896dd8f6b5ebd297392f3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 19:57:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/13689fc0816d49f6be478f291c9a96ec : SUCCESS in 37m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/90f1eda3b3fb48a4ad8a77a697173281 : SUCCESS in 50m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"017943f0dc107cb9e2e3e842160f35f06f8331d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 20:26:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/357dc89148df4ae286cf2ec5ec400f7a : SUCCESS in 13m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebe55e52fd18427da09484167b57fab1 : SUCCESS in 7m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/004daac48d5c416e8ec7cbe0f828b748 : SUCCESS in 11m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/886579336a1545d084ca6c147c4a16fb : SUCCESS in 13m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f89545d4e3646c5915d5d60049471fb : SUCCESS in 11m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/d9d2c05f099346c8ac66eaf8ae546f29 : SUCCESS in 50m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/992edbbc425f43a392c97a6cf351997a : SUCCESS in 1h 22m 20s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/13dae9c4fe6e4b9b836c55d5c204a78f : SUCCESS in 24m 29s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/750a6117e708438db72add63a292affd : SUCCESS in 31m 58s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c85b560d58c4471942662693211746f : SUCCESS in 32m 39s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/a8e88d98d88b473ea088c976f70eb694 : SUCCESS in 23m 19s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/2394d8e24ef741b9803abb4b42bd40d8 : SUCCESS in 17m 41s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/47c20292b7df4604b5ba5e5541ce9b88 : SUCCESS in 35m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0a4e5a82e344aa384ad004294e6a223 : SUCCESS in 55m 46s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/b34fba811afd456fb37b6dc9acbd76f6 : SUCCESS in 34m 18s","accounts_in_message":[],"_revision_number":2},{"id":"7f6be012feec870a48c843ad194d34a5e520f68f","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-02-04 16:55:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fc2fed4662a7bc87ba7667f4e387f13243327da4","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-02-04 16:55:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3f4144106c4722b9df55b5befcb4d7fca5768c58","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-02-04 16:56:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e240c271c9acb7259a6d9b7fbf4052e44d86675f","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-02-08 14:14:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f8a77cd8d0c37c8062cc0ac908f2d0def581f924","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 14:23:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e88e3a8adb64eeae1fdca55fcd752b323952f3fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 15:57:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0224f935198a4bd1aaa3aa886bedd9cc : SUCCESS in 8m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb3ef1600a3a4c398bd409830e4ac62a : SUCCESS in 12m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76c0e8681af74507b1e57d8f707804e0 : SUCCESS in 14m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ea700f65a2d4e558abdea5ffc95418e : SUCCESS in 14m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/36b75fa7dbef446e85ef0aa3c6ebe313 : SUCCESS in 47m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef8ce5e89ad14a648c5f95571d3fb7b2 : SUCCESS in 1h 28m 29s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/5e09a09d5f2d458f9a7b662e42ab971e : SUCCESS in 32m 42s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/e7693b7bcc004372b037a688ed12da76 : SUCCESS in 32m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5ec0f8b185f4da586830090705a02cd : SUCCESS in 55m 37s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/3eea9f68662345c1b01701532a95bdaa : SUCCESS in 1h 00m 21s","accounts_in_message":[],"_revision_number":2},{"id":"29d1f2b820c98788fb9b6aa3ff7b3100016fa246","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 15:57:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1553f3fcc227ae37cb21da0d476d6c1c515869ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 16:18:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7855188e9f164fb0981cb4f7583ffadb : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3e2b01d5f66c542491513d9c0db6c77f271e1d20","revisions":{"f6cb281cda83a3e7f030942ca36b12ddfab56bc9":{"kind":"REWORK","_number":1,"created":"2022-01-14 19:14:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/85/824785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/85/824785/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/824785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/824785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/824785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/85/824785/1"}}},"commit":{"parents":[{"commit":"ce8b7569dcf06792758200808f8545e3feefd2c5","subject":"sql: Squash train migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ce8b7569dcf06792758200808f8545e3feefd2c5"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-10 16:59:56.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-11 18:10:01.000000000","tz":0},"subject":"sql: Squash ussuri migrations","message":"sql: Squash ussuri migrations\n\nMake the following changes to the new \"initial\" migrations.\n\n- Drop foreign key constraints on \u0027domain_id\u0027 columns of \u0027user\u0027 and\n  \u0027identity_provider\u0027 tables (072)\n- Add \u0027authorization_ttl\u0027 column to \u0027identity_provider\u0027 table (073)\n- Add \u0027expiring_user_group_membership\u0027 table (073)\n\nThis concludes our squashing. We won\u0027t squash the 079 migrations, added in\nXena, since this would break fast forward upgrades. Future changes will instead\n(finally) focus on the switch to alembic.\n\nChange-Id: Ia1495cd4683d6631be2691e816734d01b03037a3\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f6cb281cda83a3e7f030942ca36b12ddfab56bc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f6cb281cda83a3e7f030942ca36b12ddfab56bc9"}]},"branch":"refs/heads/master"},"3e2b01d5f66c542491513d9c0db6c77f271e1d20":{"kind":"REWORK","_number":2,"created":"2022-01-19 18:50:03.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/85/824785/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/85/824785/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/824785/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/824785/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/824785/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/85/824785/2"}}},"commit":{"parents":[{"commit":"10b67ba1f5a38aa725c4c78eda79021a7d0ce381","subject":"sql: Squash train migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/10b67ba1f5a38aa725c4c78eda79021a7d0ce381"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-10 16:59:56.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-19 16:58:16.000000000","tz":0},"subject":"sql: Squash ussuri migrations","message":"sql: Squash ussuri migrations\n\nMake the following changes to the new \"initial\" migrations.\n\n- Drop foreign key constraints on \u0027domain_id\u0027 columns of \u0027user\u0027 and\n  \u0027identity_provider\u0027 tables (072)\n- Add \u0027authorization_ttl\u0027 column to \u0027identity_provider\u0027 table (073)\n- Add \u0027expiring_user_group_membership\u0027 table (073)\n\nThis concludes our squashing. We won\u0027t squash the 079 migrations, added in\nXena, since this would break fast forward upgrades. Future changes will instead\n(finally) focus on the switch to alembic.\n\nChange-Id: Ia1495cd4683d6631be2691e816734d01b03037a3\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3e2b01d5f66c542491513d9c0db6c77f271e1d20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3e2b01d5f66c542491513d9c0db6c77f271e1d20"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
