)]}'
{"id":"openstack%2Fkeystone~824788","triplet_id":"openstack%2Fkeystone~master~I8aa811ea336e9f613300bc21125e7582010cf5a5","project":"openstack/keystone","branch":"master","topic":"bp/remove-sqlalchemy-migrate","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-02-08 15:58:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8aa811ea336e9f613300bc21125e7582010cf5a5","subject":"sql: Remove dead code","status":"MERGED","created":"2022-01-14 19:14:29.000000000","updated":"2022-02-08 16:25:15.000000000","submitted":"2022-02-08 15:58:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824788-bp/remove-sqlalchemy-migrate","meta_rev_id":"9814295eb0fbf76bcd6eca9c5c0dfcca2a6583bf","_number":824788,"virtual_id_number":824788,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-08 15:58:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2022-02-04 16:48:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2022-02-04 16:32:53.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-04 16:53:20.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-04 16:53:20.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-14 21:09:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-04 16:32:53.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2022-02-04 16:48:18.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2022-02-04 16:53:20.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"}],"messages":[{"id":"940334347b0d1840e140eadc91adfd58c4952ab7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-14 19:14:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"695c9c13fe8482b79fce0115a8e846006052d8c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 21:09:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3274dc305d344da9bf1882e945a3c717 : SUCCESS in 21m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbc0f1429c6640bf8f7d140e219bda2f : SUCCESS in 9m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5ec97c1e1c647eda2da3b4bdfcfad70 : SUCCESS in 21m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a25f6d2fa8ec4c72b7ab10f0163e732f : SUCCESS in 17m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f02de9a1719d4ddebb541d3d27682023 : SUCCESS in 16m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/6d48b59a762f4a6a8bc4e35d091ce0a3 : SUCCESS in 59m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/844272a2da5348249c99e576d7c9317c : FAILURE in 26m 38s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/4d877bbc172e459a817430168fe2dffc : FAILURE in 27m 08s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/b3eec55229fe498b8e04433cdbff571d : FAILURE in 37m 38s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ae70d0a45934ac78162793d7d6b9504 : FAILURE in 26m 46s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/8b076a5f15f1445486cf7b2ca13a219d : FAILURE in 24m 36s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/45356829229546ec97683d47935575a8 : POST_FAILURE in 19m 13s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/43e183589bfc4cd986460816b0bc659c : FAILURE in 26m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/edbaadcfcc5c42bdac6f3861965d3d73 : FAILURE in 24m 59s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/6f0507dbc6554c6f8fa3cbecd5a56fcd : FAILURE in 26m 19s","accounts_in_message":[],"_revision_number":1},{"id":"b5447fba5e4f55375263fdf93d6f34b0a55a37bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 21:16:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f11c0fe76d02475f9b83a8076c906744 : SUCCESS in 33m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a0643771943c46538b531a92921d54d7 : SUCCESS in 32m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"002e2742ff86153821f03d5c52054f09d88df99c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-19 18:50:03.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"fb09a0eabd7c90d4d90bd7c77cca6f5e1b55e0ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-19 19:36:52.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"67909478a0a67226f30c6fadd0ef06b32269fd30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 22:11:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77baaf1356f646e399179519d99a2166 : SUCCESS in 38m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4a2191b5f26f422290421fdc35428970 : SUCCESS in 35m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7ccd0f7fc686e1e6068dfe6e22c59b5ef119869a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 23:34:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/def567664e3045d3af59e9b3d4671fd8 : SUCCESS in 17m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99cd9532edae458f9535725735fc7285 : SUCCESS in 8m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee9e7214835e45318a1bd5f6dc0261b2 : SUCCESS in 14m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51521b6de72b4e1bb5f42f6e5eb20ea2 : SUCCESS in 15m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65f9be4855744459b1a1844c1403ff6c : SUCCESS in 16m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/85455f9b6e094eada7d2f822f0f1a4e5 : SUCCESS in 1h 03m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e79012f3e0ae48e49a2d8179df70c2fe : SUCCESS in 2h 00m 19s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/c0c70c514ba04013b98b02c5a2c2c3f7 : SUCCESS in 40m 12s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/9181cf849c5e4931b52b0ea11c59b234 : SUCCESS in 44m 23s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/546b3990f7bf491a97d927dca9e2e3c5 : SUCCESS in 37m 35s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/fc6672baee514987b4ed83803e5d9efe : SUCCESS in 41m 09s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d42aa9593a084cec9155fe4429b025c8 : SUCCESS in 22m 01s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/0dfea0cfcdc24c029ad02e34ff02a554 : SUCCESS in 44m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/157b0c6ef3d247a68ffb08fc332c6b79 : SUCCESS in 55m 56s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/1f99fc0f4e46441885760a3fba05d1df : SUCCESS in 49m 38s","accounts_in_message":[],"_revision_number":3},{"id":"305c5e2baa524379c8e7ceacf827fc30ef7503fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-21 17:46:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ebc405dcfde37e4dc87b8edc32ec3e1869684e86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 18:58:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/76213eeee4c34e2187144ca0c356e335 : SUCCESS in 44m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b165fd6a3b4c4f798edd14bcee8863a0 : SUCCESS in 55m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b562720eb88085697b2ba73f65a2e238da493991","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 21:09:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad92b84275214b44962618021b2438ce : SUCCESS in 14m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dbad00189d74bdfb897dff22f3317d0 : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1a05cd8b89846a28abc66fcf52e6faa : SUCCESS in 13m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/630bfe03ab2e4397b83ad1f815d9b721 : SUCCESS in 15m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e73cfbcf848b4fd99c5e7db755279d5e : SUCCESS in 12m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/f1d8ce35ac6b470f9668a784ce2b3ec9 : SUCCESS in 1h 03m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e7324630130c46c5be6ad4496a42a9ff : SUCCESS in 1h 35m 34s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/fc3d1322a8774758a3a8d16958b5f50a : SUCCESS in 34m 50s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/4910814b56a441ada0e0be3eb86a205c : SUCCESS in 32m 58s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2edba158440843d7ba807a4ef39affef : SUCCESS in 41m 18s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/b8fc3b94a6574c8d9250877c39517822 : SUCCESS in 41m 44s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/eb32ac2d887e479cb490cdd84257e4af : SUCCESS in 17m 07s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/79775523fb874c01a3ac5a7348f7f8bf : SUCCESS in 37m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a350076a359546359a3b07abad6c1981 : SUCCESS in 46m 46s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/83c988dd2ab642b9970cc3c82bb93b34 : SUCCESS in 54m 22s","accounts_in_message":[],"_revision_number":4},{"id":"ec6b061bcd9298f66a67e8cb0762e306c4e40c37","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-02-04 16:32:53.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"c565fec450ac869f3752fdd1b43bc34f4fed95b8","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-02-04 16:48:18.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"23a81a3b84ba9181b0b1df5bac4e67c7447c7c32","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-02-04 16:53:20.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d7b3e9d21fb0483f2d0462f414e287e5e94b6f9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 14:23:57.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"69ba8b5ce58b80bf622943103e78d78417b628c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 15:58:12.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d06cf131b9ec4be88d0fc72312c352e0 : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e62e7a45ce8b4915af4ca8c2af26ac54 : SUCCESS in 14m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65f65526f2bf471f814440e218e46796 : SUCCESS in 14m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a94cee658cc43858ee211c2137e392d : SUCCESS in 15m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/e64aafaca20146d3a56b53ea621f395a : SUCCESS in 56m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d71704c8d2c9421b9d562557f7645d8c : SUCCESS in 1h 19m 00s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f1ba5aa9da774f4196bb37fb068c4097 : SUCCESS in 24m 24s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/f2bbd18b874d4ea6a16c778c2194670d : SUCCESS in 29m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e435b8395a174e79aa83ff9e8391c3bc : SUCCESS in 44m 36s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/205fa200b3bc45d1a77e2377b50a9376 : SUCCESS in 36m 19s","accounts_in_message":[],"_revision_number":4},{"id":"5940754e3102d6c804cdacf2aae6425b874834c6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 15:58:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"9814295eb0fbf76bcd6eca9c5c0dfcca2a6583bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 16:25:15.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b726990f9f348118e2822e8acce9276 : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"15847926ef61780a94b8a0f28c665aac168a77f7","revisions":{"481e5a281be95682a66c3c97dfcc7f19ac9b8055":{"kind":"REWORK","_number":1,"created":"2022-01-14 19:14:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/824788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/88/824788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/824788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/824788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/824788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/88/824788/1"}}},"commit":{"parents":[{"commit":"2a6694ea545fec0d0d71300e81668491c0f21c51","subject":"cmd: Remove deprecated \u0027--extension\u0027 argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2a6694ea545fec0d0d71300e81668491c0f21c51"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-13 18:08:28.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-14 18:22:29.000000000","tz":0},"subject":"sql: Remove dead code","message":"sql: Remove dead code\n\nChange I41584c652ab34a267009136ee2a2e159ee2f2a6e (\"sql: Squash mitaka\nmigrations\") removed the final users of this code.\n\nChange-Id: I8aa811ea336e9f613300bc21125e7582010cf5a5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/481e5a281be95682a66c3c97dfcc7f19ac9b8055"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/481e5a281be95682a66c3c97dfcc7f19ac9b8055"}]},"branch":"refs/heads/master"},"e4fb78f01face124e9bd78b60950abdf7cf9230f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-01-19 18:50:03.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/824788/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/88/824788/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/824788/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/824788/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/824788/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/88/824788/2"}}},"commit":{"parents":[{"commit":"4bb6cceba0afae8bf70ed43dc2c94401ceea17bc","subject":"cmd: Remove deprecated \u0027--extension\u0027 argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4bb6cceba0afae8bf70ed43dc2c94401ceea17bc"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-13 18:08:28.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-19 16:58:16.000000000","tz":0},"subject":"sql: Remove dead code","message":"sql: Remove dead code\n\nChange I41584c652ab34a267009136ee2a2e159ee2f2a6e (\"sql: Squash mitaka\nmigrations\") removed the final users of this code.\n\nChange-Id: I8aa811ea336e9f613300bc21125e7582010cf5a5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e4fb78f01face124e9bd78b60950abdf7cf9230f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e4fb78f01face124e9bd78b60950abdf7cf9230f"}]},"branch":"refs/heads/master"},"ddcccfdc03ceb955b6f9d392b4cd0f10a4765514":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-01-19 19:36:52.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/824788/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/88/824788/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/824788/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/824788/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/824788/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/88/824788/3"}}},"commit":{"parents":[{"commit":"72e4f98365aa5d60d6bb3176f303afdf05d03acb","subject":"cmd: Remove deprecated \u0027--extension\u0027 argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/72e4f98365aa5d60d6bb3176f303afdf05d03acb"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-13 18:08:28.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-19 19:32:54.000000000","tz":0},"subject":"sql: Remove dead code","message":"sql: Remove dead code\n\nChange I41584c652ab34a267009136ee2a2e159ee2f2a6e (\"sql: Squash mitaka\nmigrations\") removed the final users of this code.\n\nChange-Id: I8aa811ea336e9f613300bc21125e7582010cf5a5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ddcccfdc03ceb955b6f9d392b4cd0f10a4765514"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ddcccfdc03ceb955b6f9d392b4cd0f10a4765514"}]},"branch":"refs/heads/master"},"15847926ef61780a94b8a0f28c665aac168a77f7":{"kind":"REWORK","_number":4,"created":"2022-01-21 17:46:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/824788/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/88/824788/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/824788/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/824788/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/824788/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/88/824788/4"}}},"commit":{"parents":[{"commit":"9f42c5ad6a5122c2122a239876de95bff5732a2e","subject":"cmd: Remove deprecated \u0027--extension\u0027 argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9f42c5ad6a5122c2122a239876de95bff5732a2e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-13 18:08:28.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-21 13:39:30.000000000","tz":0},"subject":"sql: Remove dead code","message":"sql: Remove dead code\n\nChange I41584c652ab34a267009136ee2a2e159ee2f2a6e (\"sql: Squash mitaka\nmigrations\") removed the final users of the \u0027get_constraints_names\u0027,\n\u0027add_constraints\u0027 and \u0027remove_constraints\u0027 functions.\n\nChange I96cab42cfcfd3e86b53f25abf4cf4043af3b5667 (\"sql: Squash ocata\nmigrations\") removed the final use of the \u0027USE_TRIGGERS\u0027 constant.\n\nChange I59882d88fe593ec1ae37415b2157584f7f3c85f8 (\"sql: Remove legacy\n\u0027migrate_repo\u0027 migration repo\") removed the final use of\n\u0027_assert_not_schema_downgrade\u0027 function.\n\nChange-Id: I8aa811ea336e9f613300bc21125e7582010cf5a5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/15847926ef61780a94b8a0f28c665aac168a77f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/15847926ef61780a94b8a0f28c665aac168a77f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
