)]}'
{"id":"openstack%2Fkeystone~825378","triplet_id":"openstack%2Fkeystone~master~Ib25c75a99794a04b6549e6b5184a2029955befc1","project":"openstack/keystone","branch":"master","topic":"bp/remove-sqlalchemy-migrate","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-02-09 21:11:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib25c75a99794a04b6549e6b5184a2029955befc1","subject":"sql: Vendor \u0027oslo_db.sqlalchemy.migration\u0027","status":"MERGED","created":"2022-01-19 18:50:03.000000000","updated":"2022-02-09 21:14:10.000000000","submitted":"2022-02-09 21:11:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"825378-bp/remove-sqlalchemy-migrate","meta_rev_id":"a601a8cd6aecef2e6e107d074d49378463e090c9","_number":825378,"virtual_id_number":825378,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-09 21:11:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2022-02-09 18:00:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-04 18:00:20.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2022-02-09 18:00:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-19 22:45:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-19 23:17:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-21 16:07:45.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"CC"},{"updated":"2022-02-04 18:00:20.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2022-02-09 18:00:10.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"}],"messages":[{"id":"96da070725b9a6bcefa3ac54a8d56334e59f63ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-19 18:50:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7d2b19c2ae5a18a0a24bb561c1dafd7174a266a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-19 19:36:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7be4b471f5e95b746e591ca91b83ce31b2a76efb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 22:45:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/da2fcdbe3127412a8f0f69603f3be327 : TIMED_OUT in 1h 01m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4af8b16b13774aedb8a9e9f81dc5f98a : TIMED_OUT in 1h 01m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7ab401f34d70f05c083fd1e3694a388c4aa4745a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 23:17:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe41c9fbbf0e45549792d94438027ee0 : FAILURE in 17m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a37408d7efcd4a3396901f9d97551517 : SUCCESS in 9m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/781125c7fc294fd3872f2b3b92a3c668 : FAILURE in 15m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c72c89c0ec454eea997809925c673f6b : FAILURE in 15m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87b8b6047dbf4a32b0f938230a3a0a52 : SUCCESS in 15m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/694a41493a0b483ab176558ba01ac13c : SUCCESS in 1h 01m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/07f7555f2c4f45d4b4b97e2c691b89ff : SUCCESS in 1h 44m 31s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/7babed492f0d403f8c6de1b92a39090d : SUCCESS in 38m 31s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/438e14dda9724507981c6b335aeb1a2b : SUCCESS in 45m 03s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6cc738a8afb42dd88723d17994f932a : SUCCESS in 36m 32s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/57bccfd9acb34a85a98b61f3cff87f48 : SUCCESS in 36m 26s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/4cc71fb137024eca89186640bbaa5cf8 : SUCCESS in 21m 29s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/f8399d5859c440bea31913df5859b52d : SUCCESS in 42m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f131065a2bdf40669558d046ba500659 : SUCCESS in 54m 25s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/32da92c73786435cb41ac1899f86d3b5 : SUCCESS in 43m 17s","accounts_in_message":[],"_revision_number":2},{"id":"4d278f18b2b5499daad971ebb9379c22d98f8ccc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-20 09:52:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3378d11580ea731663c3e0f18f5750198a511de0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 11:22:25.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c7778461cf440e381536d7653de8ed6 : SUCCESS in 15m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd848dba2c464793a15ae4e2d77939c5 : FAILURE in 8m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5d3ec654c7e4ec8a576bd797142c4dc : SUCCESS in 16m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ca2e945f0614282bd7ea612d1f2d3c1 : SUCCESS in 13m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ac5bf22a70147d0bdaed9e587815a30 : SUCCESS in 12m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/2942895cc4124ae393c5d274cecadcd3 : SUCCESS in 1h 00m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bef59412747348dc98abf4d79746721d : SUCCESS in 1h 25m 40s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/384fa1d9a7244b8b81fffbb87d7535b8 : SUCCESS in 34m 49s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/d3874d94fc4b40149e979a1dcbe29c55 : RETRY_LIMIT in 3m 58s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3a6da59f49d45bd8fcf30df2109c46c : SUCCESS in 34m 25s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/cdb9247babbd4ae09cc691d15f9b74ec : SUCCESS in 34m 30s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/ffbf804fddd74b6a8667995686ee47fc : SUCCESS in 19m 43s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/bf13e677162b45019269e16d829cfa20 : SUCCESS in 43m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ab544104ce3491b90cb729df3352271 : SUCCESS in 54m 58s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/bdc414e22ac846809ba122453bea8352 : SUCCESS in 43m 11s","accounts_in_message":[],"_revision_number":3},{"id":"a98d7b8ce54f409f81713e1c85ceccc2258a0901","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 12:13:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab6e60be28e1494e8192bcfa509ac93e : TIMED_OUT in 1h 01m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cdfcb661259544fc98bf9b31038bacc0 : TIMED_OUT in 1h 02m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"883fc1ffc9b13060f144140eb242aaf39fd8658f","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-01-21 16:07:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c8c57d3b7645cc701bc1760f258cce93b0e58027","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 16:39:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/023c673abc1b407eb5440322e9fc1b5c : SUCCESS in 23m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a498924bb2041088f31aeccfb6f949d : SUCCESS in 24m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"94319e47ea92b079b0a02257b48374748a35e9c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 17:37:31.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec0b6687dc4242fd8b44601f505896c8 : SUCCESS in 16m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a0b024592014a6b928ad70e10721106 : FAILURE in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9327d8d100e406694cb18216c9bbf77 : SUCCESS in 14m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/672b1869857b452da33210d87b5a125c : SUCCESS in 13m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de0f03db53104f8a8aed0c76cf6e0bc6 : SUCCESS in 12m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/8be4f8f3820f4e21b1e3d67686df91f8 : SUCCESS in 46m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85f92eaebfd94bd386cf01c327ad7b83 : SUCCESS in 1h 25m 39s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/2bc688da3f3b4320a2acd88026a2d25e : SUCCESS in 29m 21s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/a0bc230d15d8469cb38de63f4854a847 : SUCCESS in 32m 29s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/779086eaee434bf79cc871966df3533c : SUCCESS in 34m 27s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/afd4343c46874ed282203e2290077e0d : SUCCESS in 30m 22s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/b9833a8c41474d22a87fafa8821a6405 : SUCCESS in 16m 01s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/24429c920e7a404aae61a751edfac809 : SUCCESS in 43m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3260b0effc9b4a1fbb5491badf7d4c59 : SUCCESS in 46m 08s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/37360c7842464de5a152f54edb82b48e : SUCCESS in 41m 06s","accounts_in_message":[],"_revision_number":3},{"id":"99a180ccfd6a34bfbb3a00fff68e5a64fcdda344","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-21 17:46:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"82c7c741549486e163515973ff41d75644f6448c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 18:40:39.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c92db94d9a24b4aba9cc9a1c4547c40 : FAILURE in 16m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9f9bebefd2b40928577f588adbfe005 : FAILURE in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74a704154a37439aa5d0a7c4165e4774 : FAILURE in 23m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1846d76c40c427693f942f4095e5e30 : FAILURE in 16m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ebe3f6e4f924e2ea98c27c07dbcd9a9 : SUCCESS in 13m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/264e3d7e71d94f49af227f331d5a7818 : FAILURE in 47m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4869d3458f334a9789c8fd5404673839 : FAILURE in 24m 49s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/e5368f33e3354cda89ad21168d7a5412 : FAILURE in 14m 47s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/18782393944f44c58ca813ada90b8382 : FAILURE in 26m 35s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48fd9ab239da4338baca8179ad0ec63b : FAILURE in 22m 02s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/2cae88d934ec4716ae5db5c7d738a690 : FAILURE in 16m 41s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/bcf511ea7c934c7f84ba7df12f9e32a5 : POST_FAILURE in 19m 08s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/e00f86483b6548d1ba3e57817dc34061 : FAILURE in 19m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fa6635bf2dc4dabbb2d13f3d536a403 : FAILURE in 25m 40s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/76a1a561e9f1411bafade160c89f2d27 : FAILURE in 20m 29s","accounts_in_message":[],"_revision_number":4},{"id":"0c0007eac096bf3a391fd6a99fb5391d5669c8f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 19:11:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c070c8a24d414b1c8cf9fd867cf0448b : FAILURE in 41m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a28bbe3e7d2147f28d3efd271116e015 : FAILURE in 56m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1f12617feaca240f4b6e196a642c97b3418075c6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-24 15:52:18.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2bb6a98c1b8e978d3d6e9cf7590b96634335878a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-24 18:42:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"50e97f425a2076da0d138add413d294e93e28b85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 19:44:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9063bab86c4e4ade97ae27a0c7b01682 : SUCCESS in 53m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c335ccace7764d22895cad0068d50b41 : SUCCESS in 56m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2a693b69bd89c3554d819b7ce9cdb7120879f11b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 19:59:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8de0152085db4d0294eb0e74c35cd3b2 : SUCCESS in 13m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dfe41030de64d74ad4a7579a7cc9aa4 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27709bd812d1400384bf01d3c6067267 : SUCCESS in 16m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0d7ac12e7e440f7bb863619f24f60fc : SUCCESS in 12m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de177adc386446828e121df67674be6f : SUCCESS in 13m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/8f1a8cafe41e4401b8d337f0efeb3e6e : SUCCESS in 1h 01m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa7d48fd67a64f8a8e21cdf1cfe9ea37 : SUCCESS in 1h 11m 20s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/df9fdc5e2b4e46f98534dab74202f12e : SUCCESS in 33m 55s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/dc6f423beef748329fc4e37b44f62495 : SUCCESS in 40m 05s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c6296de2a33466e98725b9b3771d01d : SUCCESS in 35m 46s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/12c619ebcc544640ba1d1940ae4764ec : SUCCESS in 28m 39s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/4405d7992bd84f4daf641b2847e7664e : SUCCESS in 20m 03s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/369a352b52094489904caff4c0de6dc7 : SUCCESS in 35m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/caaece980dae43598815d563e6c56063 : SUCCESS in 58m 10s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/eeab3fd7c07c4087a06248d219409ba6 : SUCCESS in 38m 04s","accounts_in_message":[],"_revision_number":5},{"id":"6c591e48257ff9e35423468366ee8d9334a75c9c","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-02-04 18:00:20.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"ff16702b94c8a76437de2020f50b4afe7fce4ae8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-08 16:34:01.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4db1a93f3ae79d0b3acbe2c9c57189f890fe6b59","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-02-09 18:00:10.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"26bf3fead6c5f68ab35fc3555a17f639d649053b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 18:44:40.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"86d689569b602f416d5ba15b32c79eb77b32cb81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 21:11:28.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d958f269d9d4398ab4f75335e514d03 : SUCCESS in 10m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7919ba51d4034be9a2cbcfe6c4c41efc : SUCCESS in 16m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a15da76f09e407eb487b91cbc5fd75a : SUCCESS in 17m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ccd215194424ff5bdd721f37b052c95 : SUCCESS in 19m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/419b74152a9547f7a12c8202adadccf5 : SUCCESS in 58m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f07e916163b14e48b24d8b10b1cb756b : SUCCESS in 1h 27m 33s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/047a887e75da4b3da34d5344c688309f : SUCCESS in 36m 32s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/fb6ee45c50144f31a798bb714ac30bfb : SUCCESS in 36m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c900c071f12e4433b4b1ec9fd5588279 : SUCCESS in 56m 28s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/a7b8d03f10e444e58a7f83b417f7e339 : SUCCESS in 42m 53s","accounts_in_message":[],"_revision_number":5},{"id":"a3555b276dc0cbee995f94efb04ebec136ebef78","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 21:11:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"a601a8cd6aecef2e6e107d074d49378463e090c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 21:14:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4abf56ba22f94a4a95eda4c853fa7167 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0b906c65294ac9f0723ab9f598f7f9f4c660d667","revisions":{"dd44196171d53b22bd3391818e9b63fbbe978618":{"kind":"REWORK","_number":1,"created":"2022-01-19 18:50:03.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/78/825378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/78/825378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/825378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/825378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/825378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/78/825378/1"}}},"commit":{"parents":[{"commit":"c3059706ea704578a1bbfd17f819b21a9da91c11","subject":"sql: Add additional changes to initial alembic migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c3059706ea704578a1bbfd17f819b21a9da91c11"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-19 14:38:10.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-19 16:58:16.000000000","tz":0},"subject":"WIP: db: Vendor \u0027oslo_db.sqlalchemy.migration\u0027","message":"WIP: db: Vendor \u0027oslo_db.sqlalchemy.migration\u0027\n\nThis is deprecated and will be removed in a future release of oslo.db.\nEven without that stick to prod us, we\u0027re going to need to use some of\nthe sqlalchemy-migrate APIs and it\u0027s simpler to talk to this for\neverything rather than using oslo.db for some stuff and\nsqlalchemy-migrate for the remainder.\n\nChange-Id: Ib25c75a99794a04b6549e6b5184a2029955befc1\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/dd44196171d53b22bd3391818e9b63fbbe978618"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/dd44196171d53b22bd3391818e9b63fbbe978618"}]},"branch":"refs/heads/master"},"ecf99b398db5abc453cd0dc942bb28fd43916a5a":{"kind":"REWORK","_number":2,"created":"2022-01-19 19:36:52.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/78/825378/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/78/825378/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/825378/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/825378/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/825378/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/78/825378/2"}}},"commit":{"parents":[{"commit":"5321119bc405c680d1c82a8d2acc99ce585498fa","subject":"sql: Add additional changes to initial alembic migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5321119bc405c680d1c82a8d2acc99ce585498fa"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-19 14:38:10.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-19 19:32:54.000000000","tz":0},"subject":"WIP: db: Vendor \u0027oslo_db.sqlalchemy.migration\u0027","message":"WIP: db: Vendor \u0027oslo_db.sqlalchemy.migration\u0027\n\nThis is deprecated and will be removed in a future release of oslo.db.\nEven without that stick to prod us, we\u0027re going to need to use some of\nthe sqlalchemy-migrate APIs and it\u0027s simpler to talk to this for\neverything rather than using oslo.db for some stuff and\nsqlalchemy-migrate for the remainder.\n\nChange-Id: Ib25c75a99794a04b6549e6b5184a2029955befc1\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ecf99b398db5abc453cd0dc942bb28fd43916a5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ecf99b398db5abc453cd0dc942bb28fd43916a5a"}]},"branch":"refs/heads/master"},"61879be436c521ce3878d337b99bdc5461856fef":{"kind":"REWORK","_number":3,"created":"2022-01-20 09:52:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/78/825378/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/78/825378/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/825378/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/825378/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/825378/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/78/825378/3"}}},"commit":{"parents":[{"commit":"5321119bc405c680d1c82a8d2acc99ce585498fa","subject":"sql: Add additional changes to initial alembic migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5321119bc405c680d1c82a8d2acc99ce585498fa"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-19 14:38:10.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-20 09:52:34.000000000","tz":0},"subject":"db: Vendor \u0027oslo_db.sqlalchemy.migration\u0027","message":"db: Vendor \u0027oslo_db.sqlalchemy.migration\u0027\n\nThis is deprecated and will be removed in a future release of oslo.db.\nEven without that stick to prod us, we\u0027re going to need to use some of\nthe sqlalchemy-migrate APIs and it\u0027s simpler to talk to this for\neverything rather than using oslo.db for some stuff and\nsqlalchemy-migrate for the remainder.\n\nChange-Id: Ib25c75a99794a04b6549e6b5184a2029955befc1\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/61879be436c521ce3878d337b99bdc5461856fef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/61879be436c521ce3878d337b99bdc5461856fef"}]},"branch":"refs/heads/master"},"42fe25f97593f522347d5bbf8f97ea600000c662":{"kind":"REWORK","_number":4,"created":"2022-01-21 17:46:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/78/825378/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/78/825378/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/825378/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/825378/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/825378/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/78/825378/4"}}},"commit":{"parents":[{"commit":"aebd037f501ca3aa9e927f304382c231dacb2938","subject":"sql: Move migrations to \u0027legacy_migrations\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/aebd037f501ca3aa9e927f304382c231dacb2938"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-19 14:38:10.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-21 13:39:30.000000000","tz":0},"subject":"sql: Vendor \u0027oslo_db.sqlalchemy.migration\u0027","message":"sql: Vendor \u0027oslo_db.sqlalchemy.migration\u0027\n\nThis is deprecated and will be removed in a future release of oslo.db.\nEven without that stick to prod us, we\u0027re going to need to use some of\nthe sqlalchemy-migrate APIs and it\u0027s simpler to talk to this for\neverything rather than using oslo.db for some stuff and\nsqlalchemy-migrate for the remainder.\n\nChange-Id: Ib25c75a99794a04b6549e6b5184a2029955befc1\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/42fe25f97593f522347d5bbf8f97ea600000c662"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/42fe25f97593f522347d5bbf8f97ea600000c662"}]},"branch":"refs/heads/master"},"0b906c65294ac9f0723ab9f598f7f9f4c660d667":{"kind":"REWORK","_number":5,"created":"2022-01-24 18:42:13.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/78/825378/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/78/825378/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/825378/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/825378/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/825378/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/78/825378/5"}}},"commit":{"parents":[{"commit":"aebd037f501ca3aa9e927f304382c231dacb2938","subject":"sql: Move migrations to \u0027legacy_migrations\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/aebd037f501ca3aa9e927f304382c231dacb2938"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-19 14:38:10.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-24 16:03:44.000000000","tz":0},"subject":"sql: Vendor \u0027oslo_db.sqlalchemy.migration\u0027","message":"sql: Vendor \u0027oslo_db.sqlalchemy.migration\u0027\n\nThis is deprecated and will be removed in a future release of oslo.db.\nEven without that stick to prod us, we\u0027re going to need to use some of\nthe sqlalchemy-migrate APIs and it\u0027s simpler to talk to this for\neverything rather than using oslo.db for some stuff and\nsqlalchemy-migrate for the remainder.\n\nChange-Id: Ib25c75a99794a04b6549e6b5184a2029955befc1\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0b906c65294ac9f0723ab9f598f7f9f4c660d667"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0b906c65294ac9f0723ab9f598f7f9f4c660d667"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
