)]}'
{"id":"openstack%2Fkeystone~825956","triplet_id":"openstack%2Fkeystone~master~I6c22b8a2b9b3d299b8fe9954bbec8e07487abc85","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-28 21:42:14.000000000","reason":"Takashi Kajinami replied on the change"},"21420":{"account":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"last_update":"2022-01-28 21:42:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6c22b8a2b9b3d299b8fe9954bbec8e07487abc85","subject":"Fix the wrong tempest parameter to enforce scope in Keystone","status":"ABANDONED","created":"2022-01-22 17:38:55.000000000","updated":"2022-01-28 21:42:19.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"9d5ade2c4dca4c583dedd99a7b45eef235f37c93","_number":825956,"virtual_id_number":825956,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-22 19:01:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"disliked":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":1,"date":"2022-01-28 15:20:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":-1,"date":"2022-01-28 16:54:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-22 18:15:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-22 19:01:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-28 15:20:29.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2022-01-28 16:54:13.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"}],"messages":[{"id":"3ec9343d9d19c35d5830e350f8fa9a1473a98f1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-22 17:38:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b8a96ddbd6aa1143a63de02e953b44dc4f12856","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-22 18:15:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ed5fafd5c7304009bc084dbbc4b0de3b : SUCCESS in 29m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b1c05710a76d44a38ee6f43d61e74cbf : SUCCESS in 30m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"10ed533a3e51638d2d074eab74770186215be1b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-22 19:01:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aadd46c7b3264525973500a15c4e9c16 : SUCCESS in 15m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8afcc07d9ab4ffebaf70ae5c7dff453 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ab3f26688f840c1b7f46acf8f5977cc : SUCCESS in 15m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2f3bdf80a724226ace533f8e3b26b99 : SUCCESS in 13m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c78f4de6e20f48c4b5fb5362184f7c1f : SUCCESS in 13m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/d750d3b812c04d9a8ef93a536b3e04f4 : SUCCESS in 54m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2af6419a69104006b27cabc661c19a86 : SUCCESS in 1h 21m 19s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/c67777d0439a4a7483fc66ba9f0ac525 : SUCCESS in 27m 17s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/3b8d159815244ff78dace200b0f3e4c6 : SUCCESS in 35m 08s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d942d3b2dd54744aa1403451a8c4519 : SUCCESS in 28m 31s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/75622e8844b5470a965e143f987246bd : SUCCESS in 28m 53s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/7f6de9b19e9e4947b7be57f50f04cada : SUCCESS in 9m 21s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/d659d850343843b38c5fc241eb430d0f : SUCCESS in 34m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1fe37c7343b498c813ea7295a855140 : SUCCESS in 33m 05s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/9f75de2246ff49209e58fc1e3b683028 : SUCCESS in 19m 26s","accounts_in_message":[],"_revision_number":1},{"id":"8b2eead8e9ddd5d6a802c34d002f467970df0622","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-01-28 15:20:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"aea2049559b335a7f725c24cd754e79c4ee2d10a","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-01-28 16:54:13.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8e0074fc7974073accccf2db081ebc1c0cbdf11f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-28 21:42:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9d5ade2c4dca4c583dedd99a7b45eef235f37c93","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-28 21:42:19.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1153ae88d3afa994cbedf2ed3369498b27b5b460","revisions":{"1153ae88d3afa994cbedf2ed3369498b27b5b460":{"kind":"REWORK","_number":1,"created":"2022-01-22 17:38:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/56/825956/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/56/825956/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/56/825956/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/56/825956/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/56/825956/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/56/825956/1"}}},"commit":{"parents":[{"commit":"771c943ad2116193e7bb118c74993c829d93bd71","subject":"Add \u0027WarningsFixture\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/771c943ad2116193e7bb118c74993c829d93bd71"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-22 17:38:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-22 17:38:16.000000000","tz":540},"subject":"Fix the wrong tempest parameter to enforce scope in Keystone","message":"Fix the wrong tempest parameter to enforce scope in Keystone\n\nChange-Id: I6c22b8a2b9b3d299b8fe9954bbec8e07487abc85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1153ae88d3afa994cbedf2ed3369498b27b5b460"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1153ae88d3afa994cbedf2ed3369498b27b5b460"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
