)]}'
{"id":"openstack%2Fkeystone~827915","triplet_id":"openstack%2Fkeystone~master~Id3d2769af6004bd4d4792c4488ac5eba94a5a776","project":"openstack/keystone","branch":"master","topic":"bp/remove-sqlalchemy-migrate","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-02-11 17:41:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id3d2769af6004bd4d4792c4488ac5eba94a5a776","subject":"sql: Remove dead helpers","status":"MERGED","created":"2022-02-04 19:35:15.000000000","updated":"2022-02-11 17:43:48.000000000","submitted":"2022-02-11 17:41:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827915-bp/remove-sqlalchemy-migrate","meta_rev_id":"0078fd93d18601905c83dbc971fa8c72270b3cf6","_number":827915,"virtual_id_number":827915,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-11 17:41:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2022-02-10 19:43:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2022-02-11 16:36:46.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2022-02-11 16:36:24.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2022-02-11 16:36:24.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-04 20:26:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-10 19:43:17.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2022-02-11 16:36:24.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2022-02-11 16:36:46.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"e3d25802f0b3a1f37e93e12451b0b308e006f4b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-04 19:35:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a8c7590c004ba9e12f84289a5ecc76ac609b5aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 20:26:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bbe08fd6b4b4394b2c5d1a3045dbfab : SUCCESS in 18m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f41b7b8f4a7241b59ba058483296eec8 : SUCCESS in 9m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e15fe6988af64cd492718e2271356b80 : SUCCESS in 17m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d1a675f4de348b8936a1453f0d85e91 : SUCCESS in 17m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84571e288bb149ae96bf3b6b7baebea1 : SUCCESS in 18m 15s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/27cb64b851c54b72a0bcd5cbbfc323b5 : SUCCESS in 42m 43s","accounts_in_message":[],"_revision_number":1},{"id":"cd0723062ddac50cf0cfee0ed8c16c8796b4168c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 00:21:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2060291c46e04edabbf8417c8a7d3499 : SUCCESS in 59m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/16c40545fcf142719199d509b3de1706 : SUCCESS in 57m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"94e452592b00d5d95e05bc4291b372a08bac681e","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-02-10 19:43:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3495f722eb62a40666b04ff259b9ecbb244f4aaf","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-02-11 16:36:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"127af19c2f6176ba2bbfe71bf4efa18a310fee2b","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-02-11 16:36:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1a543faa4cf43e583dbbb4aa058f2df0781b8243","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 16:37:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d29ce1d6c7068d3d29b7883e35106b0af0a8c496","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 17:41:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/173d75bee15846948c42ef2a33617d20 : SUCCESS in 9m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d41c4cf92efe4292a1fa11ccc92055de : SUCCESS in 12m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61bd72cbf2794506bf0e8954308d1a2c : SUCCESS in 14m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cc63ebdd27c47d8a3ce3dab100881aa : SUCCESS in 12m 54s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/92270b0f7b364624a5b4a800a269f5e1 : SUCCESS in 35m 13s","accounts_in_message":[],"_revision_number":1},{"id":"9fdf16a865b5008c39297fb4b9171be7b01004e4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 17:41:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0078fd93d18601905c83dbc971fa8c72270b3cf6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 17:43:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fb74301e9f04799814389c030793f32 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a6887c13ea75cbd3a09abab3dbcfe194637b3ed0","revisions":{"a6887c13ea75cbd3a09abab3dbcfe194637b3ed0":{"kind":"REWORK","_number":1,"created":"2022-02-04 19:35:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/15/827915/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/15/827915/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/15/827915/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/15/827915/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/15/827915/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/15/827915/1"}}},"commit":{"parents":[{"commit":"3b24c79331613d753e7157c58a531b707eea9161","subject":"sql: Add initial Yoga migration branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3b24c79331613d753e7157c58a531b707eea9161"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-02-04 17:35:09.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-02-04 19:23:05.000000000","tz":0},"subject":"sql: Remove dead helpers","message":"sql: Remove dead helpers\n\n\u0027assertTableCountsMatch\u0027 should have been removed in change\nI41584c652ab34a267009136ee2a2e159ee2f2a6e (\"sql: Squash mitaka\nmigrations\")\n\n\u0027does_constraint_exist\u0027 should have been removed in change\nIc499a996e1b9c6f0a9d885b4a9e655008002d49f (\"sql: Squash queens\nmigrations\")\n\n\u0027does_pk_exist\u0027 should have been removed in change\nI6beb8c9c827546757f2d7673a917b98f8b917012 (\"sql: Squash rocky\nmigrations\")\n\n\u0027assertTableExists\u0027 should have been removed in change\nI58e4b6833036a25ea8a1202c87ceebdbe932b447 (\"sql: Squash stein\nmigrations\")\n\n\u0027insert_dict\u0027, \u0027does_index_exist\u0027, and \u0027does_unique_constraint_exist\u0027\nshould have been removed in change\nIfc71efb8609bc26ce998a9fa48b207abf8b9a38b (\"sql: Squash train\nmigrations\")\n\n\u0027does_fk_exist\u0027 should have been removed in change\nIa1495cd4683d6631be2691e816734d01b03037a3 (\"sql: Squash ussuri\nmigrations\")\n\nChange-Id: Id3d2769af6004bd4d4792c4488ac5eba94a5a776\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a6887c13ea75cbd3a09abab3dbcfe194637b3ed0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a6887c13ea75cbd3a09abab3dbcfe194637b3ed0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
