)]}'
{"id":"openstack%2Fkeystone~827961","triplet_id":"openstack%2Fkeystone~stable%2Fxena~I77148641715efe09e3adc2e9432e66e50fb444b4","project":"openstack/keystone","branch":"stable/xena","topic":"fix-ldap-bytes-stable/xena","attention_set":{},"removed_from_attention_set":{"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2022-05-19 19:17:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I77148641715efe09e3adc2e9432e66e50fb444b4","subject":"Fix issue with LDAP backend returning bytes instead of string","status":"MERGED","created":"2022-02-07 09:25:50.000000000","updated":"2022-05-19 19:19:18.000000000","submitted":"2022-05-19 19:17:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827961-fix-ldap-bytes-stable/xena","cherry_pick_of_change":819477,"cherry_pick_of_patch_set":3,"meta_rev_id":"5bb9a396c9bb8e7e4ed9e9beba2b5abd03344153","_number":827961,"virtual_id_number":827961,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-19 19:17:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2022-05-19 17:41:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2022-05-19 17:41:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-07 10:29:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-07 10:43:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-19 17:41:42.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"1dd40eac6e9d96614b5b8adcd6c636ec7c2b5532","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-02-07 09:25:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d1870a7d6394609042833caccde10605dede4f8c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 10:29:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5fed641fa6b640e09da3eb6b52f104e4 : TIMED_OUT in 1h 01m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/340ab2c8b8404cbfb7ffe8c267d99e2d : TIMED_OUT in 1h 01m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"887c6a5182c60192de01db68842a10dae98f2b8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 10:43:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/472717290fdb40708a8301127d3acc28 : SUCCESS in 16m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dc43606755a45dd9629a41457e91377 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d13d123ebd1423bbc20bff6d9916a26 : SUCCESS in 17m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba957dc0d06043459f02d7b8978de288 : SUCCESS in 19m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc1e8ce86a8643028c18075243f696cc : SUCCESS in 13m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/123d3ab3f6454094ae9d240bf7e2d142 : SUCCESS in 1h 03m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c435fe828b6844d781e35dcc16cf6def : SUCCESS in 1h 15m 45s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/dc46cfb921f245f9a257fbf1459f0634 : SUCCESS in 28m 14s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/eed4f022c8614640858feec7005b794e : SUCCESS in 54m 32s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18814e26d5bb40b19b8e49cd25ac3651 : SUCCESS in 28m 08s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/e56b7885948046aaae6fdf20cd97d693 : SUCCESS in 35m 59s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/0ae291ee5259427595a74414f672de77 : SUCCESS in 11m 09s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/a868d1fbfac14bc9b01314e2d3e3c2bc : FAILURE in 14m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4640c5d4a1374018a936fe1dfc0cc29f : SUCCESS in 57m 27s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/fa3ae479d26f4adab14379efb02d249f : SUCCESS in 35m 37s","accounts_in_message":[],"_revision_number":1},{"id":"d7eafe1c89d59160d0f71c9b581019f602d3a8c3","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-05-19 17:41:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"efdc5f43d1b2540bf6c1b11ee8dd6a05eea6b0f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 17:42:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6696e0ad341d2794f1ee2141b41b06def0f5ab50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 19:17:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4611b0794759475cbd28c82d5a269048 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c679699932e454a84cd714a5497dc3a : SUCCESS in 13m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3705155d92c54dd5ac67e7a46f4d49ad : SUCCESS in 18m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66617731a6684838808c5236874a2b26 : SUCCESS in 14m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/f521239fe3c14f5495b0f60a6fd5d624 : SUCCESS in 57m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e84812a482d4b499cf42f832828091f : SUCCESS in 1h 16m 23s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/ddbcb76c00ae4ac9982f21d960fdad5e : SUCCESS in 32m 03s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/0300ba78aa934dfcb407fdda81724a64 : SUCCESS in 28m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f8a4cf7bbc5414c96613c052e44927f : SUCCESS in 44m 25s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/7823210da1b34233935724e852069313 : SUCCESS in 39m 18s","accounts_in_message":[],"_revision_number":1},{"id":"d3b903ed680bd6490ebcacbd5ea28f74175f50ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 19:17:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5bb9a396c9bb8e7e4ed9e9beba2b5abd03344153","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 19:19:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39fac3d9f50e49609c10ab9fb7dbe96a : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca93f062df0af06616eefda3f2574ea814ee53bd","revisions":{"ca93f062df0af06616eefda3f2574ea814ee53bd":{"kind":"REWORK","_number":1,"created":"2022-02-07 09:25:50.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/61/827961/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/61/827961/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/61/827961/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/61/827961/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/61/827961/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/61/827961/1"}}},"commit":{"parents":[{"commit":"20be021347664438520f420eb26168435f7b860b","subject":"Merge \"Fix typos in ec2 credential policies\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/20be021347664438520f420eb26168435f7b860b"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-11-26 14:28:27.000000000","tz":60},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2022-02-07 09:25:50.000000000","tz":0},"subject":"Fix issue with LDAP backend returning bytes instead of string","message":"Fix issue with LDAP backend returning bytes instead of string\n\nWhen connecting to some LDAP server software, the ldap client returns\nbytes instances instead of the expected strings. This can result in\neither being transparently converted to strings, when the data is\ninserted via sqlalchemy into the database, or could be used as\ninput to other functions, and/or cached, which causes unexpected\nresults.\n\nCloses-Bug: #1952458\nResolves: rhbz#1964872\nChange-Id: I77148641715efe09e3adc2e9432e66e50fb444b4\n(cherry picked from commit 1e0cd90191663c100c165d4c6a2b1ca796b5af25)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ca93f062df0af06616eefda3f2574ea814ee53bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ca93f062df0af06616eefda3f2574ea814ee53bd"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
