)]}'
{"id":"openstack%2Fkeystone~827964","triplet_id":"openstack%2Fkeystone~stable%2Fussuri~I77148641715efe09e3adc2e9432e66e50fb444b4","project":"openstack/keystone","branch":"stable/ussuri","topic":"fix-ldap-bytes-stable/xena-stable/wallaby-stable/victoria-stable/ussuri","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2022-06-14 12:21:53.000000000","reason":"Douglas Mendizábal replied on the change"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2022-06-14 13:38:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I77148641715efe09e3adc2e9432e66e50fb444b4","subject":"Fix issue with LDAP backend returning bytes instead of string","status":"MERGED","created":"2022-02-07 09:27:10.000000000","updated":"2022-06-14 13:40:12.000000000","submitted":"2022-06-14 13:38:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827964-fix-ldap-bytes-stable/xena-stable/wallaby-stable/victoria-stable/ussuri","cherry_pick_of_change":827963,"cherry_pick_of_patch_set":1,"meta_rev_id":"0d24ca915e10c12103b56ed1dde2a0a357f9acc7","_number":827964,"virtual_id_number":827964,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-14 13:38:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2022-06-14 12:21:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2022-06-14 12:21:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"}],"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-07 10:58:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-14 01:30:13.000000000","updated_by":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"reviewer":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"state":"CC"},{"updated":"2022-06-14 01:30:13.000000000","updated_by":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"c1e892c1fc6eca7708b2897c33062cee88ebfd44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-02-07 09:27:10.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"d9265f6aef598bcf2f551d7d9ecfc41c150654d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 10:58:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abdf2977aed64b69a16d3aba4d04d43e : SUCCESS in 18m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd0dfe77151b43bd8c37564acf499b10 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fecfd7cbf6db4d8da70edaa6ff9b716e : SUCCESS in 11m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/331e0904bbb0467bb6bf4f45a2550bac : SUCCESS in 16m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a21d713919fd4277af402ab2287925c8 : SUCCESS in 12m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/0f27f19963ef4d6ebbe1248b0c1eb756 : SUCCESS in 1h 18m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3de86768fce041cd8edc791c2c2201c1 : SUCCESS in 1h 25m 27s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/30c892483a9b4a22a53c21c1ee5d52af : SUCCESS in 29m 11s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/53a5c4e3200a45c9885647a1eaa073b8 : FAILURE in 25m 17s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/9a58784c29614c128f15ac423f11e448 : FAILURE in 30m 06s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/dd4ef2da10f149fcb2fb2651fe139495 : FAILURE in 10m 18s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ef41738d6e4147ed9e49edd0e44b3fb8 : SUCCESS in 39m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56d41f7a29c846adb873d3ab8df3c948 : SUCCESS in 1h 07m 24s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/ed5b1094f6c8416292e94e0e4bd3a648 : SUCCESS in 41m 52s","accounts_in_message":[],"_revision_number":1},{"id":"2c16834a6df765432b8fd3a2ebd7b160187c05c2","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2022-06-14 01:30:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f0d9abbfe80a4a154f1a4508ee6edbdb65225ba2","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-06-14 12:21:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5b2558f2a3a7764435662f9d78f0368f9d03704d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 12:22:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02e997bff553949c196fd5bba644187575f56f07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 13:38:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2192dcebd544455982cf7c0c3635308 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e14c25be7a2469191783f99c3e90967 : SUCCESS in 13m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c17a541e0b8243dcb12e62b9efd3ce72 : SUCCESS in 14m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bcc8bd76a034f9d82a246b5b97ede5d : SUCCESS in 13m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/0a131f8b29684ad5a6e959bd5e1d16ff : SUCCESS in 1h 01m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb32bc3b74004fa081292e0e1c510fcf : SUCCESS in 1h 15m 34s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/23f8a7c9da1a42d99ad40b10c1eda6c4 : SUCCESS in 30m 33s\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/4404d3eab07b468ba2ac865c114d0b9c : FAILURE in 38m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a12ae388f91489680dc0f2238c077b4 : SUCCESS in 49m 42s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/8c0b3ff83ef04c6a95d2291fa8fcffc6 : SUCCESS in 40m 38s","accounts_in_message":[],"_revision_number":1},{"id":"be9d431650b4926b540dd550d089dd1c74258f83","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 13:38:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0d24ca915e10c12103b56ed1dde2a0a357f9acc7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 13:40:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8efa9c89cc6a4114a5c3aec608cf60e6 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"256dbe9fe8a7b40dc8f1fd65cc004c92a5721940","revisions":{"256dbe9fe8a7b40dc8f1fd65cc004c92a5721940":{"kind":"REWORK","_number":1,"created":"2022-02-07 09:27:10.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/64/827964/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/64/827964/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/64/827964/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/64/827964/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/64/827964/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/64/827964/1"}}},"commit":{"parents":[{"commit":"7941a0b7849dee360baeb7bc9af274b34ef7fa22","subject":"Merge \"Fix typos in application credential policies\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7941a0b7849dee360baeb7bc9af274b34ef7fa22"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-11-26 14:28:27.000000000","tz":60},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2022-02-07 09:27:10.000000000","tz":0},"subject":"Fix issue with LDAP backend returning bytes instead of string","message":"Fix issue with LDAP backend returning bytes instead of string\n\nWhen connecting to some LDAP server software, the ldap client returns\nbytes instances instead of the expected strings. This can result in\neither being transparently converted to strings, when the data is\ninserted via sqlalchemy into the database, or could be used as\ninput to other functions, and/or cached, which causes unexpected\nresults.\n\nCloses-Bug: #1952458\nResolves: rhbz#1964872\nChange-Id: I77148641715efe09e3adc2e9432e66e50fb444b4\n(cherry picked from commit 1e0cd90191663c100c165d4c6a2b1ca796b5af25)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/256dbe9fe8a7b40dc8f1fd65cc004c92a5721940"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/256dbe9fe8a7b40dc8f1fd65cc004c92a5721940"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
