)]}'
{"id":"openstack%2Fkeystone~839001","triplet_id":"openstack%2Fkeystone~master~I0ac3d05389041ac58fe2347171541ffaaf151fdf","project":"openstack/keystone","branch":"master","topic":"remove-training-labs-note","attention_set":{},"removed_from_attention_set":{"33786":{"account":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"last_update":"2022-04-22 16:30:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0ac3d05389041ac58fe2347171541ffaaf151fdf","subject":"Remove the note of training-labs","status":"MERGED","created":"2022-04-22 08:03:27.000000000","updated":"2022-04-24 05:13:50.000000000","submitted":"2022-04-22 16:30:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839001-remove-training-labs-note","meta_rev_id":"3068f8e8bf7d24da52b583bb13b0e0c588bf67b5","_number":839001,"virtual_id_number":839001,"owner":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":33801,"name":"wang","email":"wangleic@uniontech.com"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-22 16:30:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2022-04-22 15:13:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2022-04-22 15:13:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2022-04-22 15:13:56.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2022-04-24 05:13:50.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":33801,"name":"wang","email":"wangleic@uniontech.com"},{"value":1,"date":"2022-04-22 15:13:34.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2022-04-22 15:14:07.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":33801,"name":"wang","email":"wangleic@uniontech.com"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33801,"name":"wang","email":"wangleic@uniontech.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-22 08:17:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-22 10:12:27.000000000","updated_by":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"reviewer":{"_account_id":33801,"name":"wang","email":"wangleic@uniontech.com"},"state":"CC"},{"updated":"2022-04-22 15:13:15.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2022-04-22 15:13:34.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2022-04-22 15:13:36.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2022-04-22 15:13:56.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2022-04-24 05:13:50.000000000","updated_by":{"_account_id":33801,"name":"wang","email":"wangleic@uniontech.com"},"reviewer":{"_account_id":33801,"name":"wang","email":"wangleic@uniontech.com"},"state":"REVIEWER"}],"messages":[{"id":"ac1bd28e93c04c1bb34915962054afca45050690","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2022-04-22 08:03:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"441b596350dcf96e135f19b9860e3b92f33098a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 08:17:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8aa5e1d7f984bf4ba6921a019f694b8 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33b7fcfd74934f4e88b494e2a14dd9df : SUCCESS in 11m 30s","accounts_in_message":[],"_revision_number":1},{"id":"b46fe49c153a370d8d93a5603a213412b38eeb33","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-04-22 15:13:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3c916d9be86cd7da71a54be9389f33d2e5624f46","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-04-22 15:13:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6fc4239a62560a88c21625441531e66e8b9a8982","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-04-22 15:13:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4b96ec0984883ae5f0ae04113084713d6ede0f02","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2022-04-22 15:13:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9fe76c1542762ad4bee22d9610c8a0eb70d741e9","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2022-04-22 15:14:07.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f0c2e4f8e11d19209bcc6d8ea9aa59c2d28d169b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 15:14:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"66ab32a36c67cd68bfe6b79a32ad16b276b2e781","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 16:30:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61875edc4c064ceca41ac559548ac329 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5b9d33dec684fa09c07a7176962a4c8 : SUCCESS in 12m 12s","accounts_in_message":[],"_revision_number":1},{"id":"f646afb4e81e3981f52b259a5d6600365c6132d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 16:30:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5afbff5ade000832a72ffb927059c2507f20220e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 16:31:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63b59444c146499db79d26979d787172 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1},{"id":"3068f8e8bf7d24da52b583bb13b0e0c588bf67b5","author":{"_account_id":33801,"name":"wang","email":"wangleic@uniontech.com"},"date":"2022-04-24 05:13:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"644deedda5fc721bc7490113e86b30145c96f2af","revisions":{"644deedda5fc721bc7490113e86b30145c96f2af":{"kind":"REWORK","_number":1,"created":"2022-04-22 08:03:27.000000000","uploader":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"ref":"refs/changes/01/839001/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/01/839001/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/839001/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/839001/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/839001/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/01/839001/1"}}},"commit":{"parents":[{"commit":"c141e6164e77f5ebb1369016d06de23a308ef510","subject":"trivial: Fix typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c141e6164e77f5ebb1369016d06de23a308ef510"}]}],"author":{"name":"Han Guangyu","email":"hanguangyu@uniontech.com","date":"2022-04-22 07:15:21.000000000","tz":480},"committer":{"name":"Han Guangyu","email":"hanguangyu@uniontech.com","date":"2022-04-22 07:15:21.000000000","tz":480},"subject":"Remove the note of training-labs","message":"Remove the note of training-labs\n\nTraining-labs had been officially retired as no maintainer.\nThe information of training-labs has been deleting in the openstack\ndocumentatioan. It is not appropriate to continue the presentation in\nnote form here.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-October/025586.html\n[2] https://opendev.org/openstack/training-labs/commit/e78d74f10558ab3e6a9a6fd7d45e617c15e9c3d8\n\nChange-Id: I0ac3d05389041ac58fe2347171541ffaaf151fdf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/644deedda5fc721bc7490113e86b30145c96f2af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/644deedda5fc721bc7490113e86b30145c96f2af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
