)]}'
{"id":"openstack%2Fkeystone~839011","triplet_id":"openstack%2Fkeystone~stable%2Fqueens~I0ac3d05389041ac58fe2347171541ffaaf151fdf","project":"openstack/keystone","branch":"stable/queens","attention_set":{},"removed_from_attention_set":{"33786":{"account":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"last_update":"2022-07-15 16:49:53.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0ac3d05389041ac58fe2347171541ffaaf151fdf","subject":"Remove the note of training-labs","status":"ABANDONED","created":"2022-04-22 08:16:42.000000000","updated":"2022-07-15 16:49:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3dd2ce7569b587b91e8d5835a7c41df88906c0cd","_number":839011,"virtual_id_number":839011,"owner":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-04-22 08:26:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2022-04-22 15:14:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-22 08:26:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-22 15:14:04.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"e633189f6371576a4032a26e946c0c49f005aa96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2022-04-22 08:16:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d30658455b72b86d076096a1589fc9cc7429b9df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 08:26:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c28017fb1ae64e9882967481a166f20d : SUCCESS in 7m 40s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/29217c2b95a54533a6f9e147f4150119 : SUCCESS in 8m 10s","accounts_in_message":[],"_revision_number":1},{"id":"399a07564b8622a653281afe1520725e65e61be3","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-04-22 15:14:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3dd2ce7569b587b91e8d5835a7c41df88906c0cd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-07-15 16:49:53.000000000","message":"Abandoned\n\nAbandoning unmerged stable/queens changes.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9266a83d7029eed2f0a7c360b6798db8da7e9f9c","revisions":{"9266a83d7029eed2f0a7c360b6798db8da7e9f9c":{"kind":"REWORK","_number":1,"created":"2022-04-22 08:16:42.000000000","uploader":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"ref":"refs/changes/11/839011/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/11/839011/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/11/839011/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/11/839011/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/11/839011/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/11/839011/1"}}},"commit":{"parents":[{"commit":"36e65cc4ca674d61f2dde6b16e310886fcf10a06","subject":"Merge \"Retry update_user when sqlalchemy raises StaleDataErrors\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/36e65cc4ca674d61f2dde6b16e310886fcf10a06"}]}],"author":{"name":"Han Guangyu","email":"hanguangyu@uniontech.com","date":"2022-04-22 07:15:21.000000000","tz":480},"committer":{"name":"Han Guangyu","email":"hanguangyu@uniontech.com","date":"2022-04-22 08:00:12.000000000","tz":480},"subject":"Remove the note of training-labs","message":"Remove the note of training-labs\n\nTraining-labs had been officially retired as no maintainer.\nThe information of training-labs has been deleting in the openstack\ndocumentatioan. It is not appropriate to continue the presentation in\nnote form here.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-October/025586.html\n[2] https://opendev.org/openstack/training-labs/commit/e78d74f10558ab3e6a9a6fd7d45e617c15e9c3d8\n\nChange-Id: I0ac3d05389041ac58fe2347171541ffaaf151fdf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9266a83d7029eed2f0a7c360b6798db8da7e9f9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9266a83d7029eed2f0a7c360b6798db8da7e9f9c"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
