)]}'
{"id":"openstack%2Fkeystone~841636","triplet_id":"openstack%2Fkeystone~master~I070d034926507f9b74c448c8cfa91f66ac568adf","project":"openstack/keystone","branch":"master","topic":"html-rebase","attention_set":{"2218":{"account":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"last_update":"2022-05-12 22:05:02.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I070d034926507f9b74c448c8cfa91f66ac568adf","subject":"Change Header to Basic auth on authentication failure.","status":"NEW","created":"2022-05-12 20:39:32.000000000","updated":"2024-09-13 20:01:06.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"994f4871bc94d69e78d72a97760a8c683e935497","_number":841636,"virtual_id_number":841636,"owner":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-09-13 20:01:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-12 21:26:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-12 22:05:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e69a39b2e816081dcd00fee0618c27a18cd374f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2022-05-12 20:39:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88f6621c6710cd5a613b70c27f5995ec8f315687","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 21:26:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a51b458701e4c75b3d0fbf7fa829bcf : SUCCESS in 42m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ea0149c201f6475a80c26049c4d0965a : SUCCESS in 44m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e3db8ebc63c706fe1eeefa6bc2feb1d516d1070a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 22:05:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cae86b03567c442fb90aca5d224d6914 : SUCCESS in 15m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dae00deb2b44304bbad7d154b6be36f : FAILURE in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94179e4c5ed94850920d99ca4a60b391 : SUCCESS in 12m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89d742ca61694588bc659bf5537e0252 : SUCCESS in 11m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3cab39d4e7da4f068f4815ca2335d245 : SUCCESS in 12m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/749b0816e120414ebcbfa01a4a084dc4 : FAILURE in 8m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/0c2d4fe02dd643298dc343d28790ed3b : SUCCESS in 51m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8f5054d98e5406d95cc0a32ca6c5bf8 : SUCCESS in 1h 20m 55s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/58a1f20385ec48e69fe3a007593ff98f : SUCCESS in 37m 11s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/3ca9fd136291445193cb56a1c4f2561a : FAILURE in 16m 20s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e79129701209491a8f4b5a3a6440dd50 : SUCCESS in 28m 48s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/62420e60202748c59ee9a8834438d19b : SUCCESS in 28m 57s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/e0e85c9309014c969bb83b9128649384 : SUCCESS in 13m 44s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/dce1b772d3cf4ba6a4bddff2f8b3fa5f : SUCCESS in 33m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a6baf0754c340578e81d48579a35702 : SUCCESS in 39m 14s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/b4863bf43e6143a4948ae36310e3157a : SUCCESS in 37m 09s\n\nWarning:\n  Comments left for invalid file keystone/api/discovery.py\n  Comments left for invalid file keystone/common/json2html.py","accounts_in_message":[],"_revision_number":1},{"id":"77ee2b2db1269cfa022806780ae9641299a6e17d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2024-08-21 18:06:44.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9a1ed6143469e430ba5767c99efc8c0571adb65a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 19:14:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf703c2107d74d049c85a7dbb5c5a35c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ddeca18b5474fa199e01ff60b4b7ae6 : FAILURE in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88dedb597e2645cca08ef30127750693 : FAILURE in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37d03e6fcef84295b6d07374eb69501a : FAILURE in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/052436ab17714f82a0d2d34dc74fb1db : FAILURE in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e256157163d34fea8b904ab7afdb06b3 : FAILURE in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80f1e65fdbbf4fccb0c0bbf61487575e : FAILURE in 6m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/0a5bd3bb3b1b4282b72f4209d76176b1 : FAILURE in 1h 00m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c6df9ba1a6fa4a219e81aba29513c5af : FAILURE in 13m 21s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/42f404139a6b4e68b0537babb61726c1 : FAILURE in 22m 11s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/f3dad6e3be124dc798832392015c5b90 : FAILURE in 32m 16s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d701e01788124c428566dd057a9405a5 : FAILURE in 20m 50s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/e69441f6ae6f4f4591830a5e0c6071c3 : FAILURE in 23m 18s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/b4b08049d19f4f48af59fa11d44c066d : POST_FAILURE in 18m 54s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/e2925e7a25714c3396d83a36d528032a : FAILURE in 25m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24e282262db04a6c893e06fba64cec10 : FAILURE in 21m 32s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/55f382ab45a040dcaab8a7c08691e269 : FAILURE in 18m 40s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/3254aff969564ded8e038ea0a8d65641 : RETRY_LIMIT in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9f623b16363a0ad545916e38d5de81b3cc10bab4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 07:59:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a860e279464d411bae0cfabe6f82e677\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/05fcb4934e0e4508ba8d0e1ca2ae702e : FAILURE in 12m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/df6b876476c349328e82592c7c9c7348 : FAILURE in 9m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"158e24fd0921312d6699488651e3e6c80e32ec23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2024-09-13 19:24:02.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"da138cb58b326d7dc2606b1e8ac006d252cc83df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 19:56:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a10e56fefeea4382812a8c9d48b2a9f8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c121175eaf0c48619f2bf94169932ec8 : FAILURE in 12m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1d8b6d52a3a74bf98627be0c611cffd2 : FAILURE in 10m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"994f4871bc94d69e78d72a97760a8c683e935497","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 20:01:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6034cb1d0d954de499e674648fcd1e77\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c32c5c6410f4ac9aa1c686bd5c01db9 : FAILURE in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22aadb8e8b634a1f85bf623a8e67e2c8 : FAILURE in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0b4cd6ca6c442f18c5ca29a47913ca9 : FAILURE in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4c9d3bfdb5524c62b16c1e5d4d1aef2b : FAILURE in 3m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/636cb89cd60942828cbd1a0d08afea74 : FAILURE in 5m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a769cd7d3ee7422f97f2c6777e633303 : FAILURE in 5m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/f41f67e9c61b4fe9a82ee74f7a23035b : FAILURE in 30m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/71249421521540c0a8f0cd3236fe5093 : FAILURE in 11m 44s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/0c875419f8aa4c71a8b774728c137066 : FAILURE in 13m 28s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/568eaa4ab1594e80b532ace4e653e8da : FAILURE in 31m 07s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92d5b9c29dfd430abdefd9a83b163774 : FAILURE in 11m 04s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/daede009b3ce4fa5a7c6966a5751e9c3 : FAILURE in 13m 28s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/3e35b3d16cf24d56847351b1aec2c7fe : POST_FAILURE in 16m 46s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/6d9eecd90acb424f9ada5b1abd8b4833 : FAILURE in 13m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a085b20eab9343adb78dcb122916c1c4 : FAILURE in 22m 32s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/af2475ddcfe4457c80381a6536e900e1 : FAILURE in 11m 44s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/d68656acaf304775978e39013d003da4 : POST_FAILURE in 5m 28s (non-voting)\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: file keystone/common/json2html.py has only 209 lines\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: file keystone/common/json2html.py has only 209 lines\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: file keystone/common/json2html.py has only 209 lines\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: file keystone/common/json2html.py has only 209 lines\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: file keystone/common/json2html.py has only 209 lines\u0027\n  Comments left for invalid file keystone/api/discovery.py\n  Comments left for invalid file keystone/common/json2html.py","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0eef0759ba3c85e465a9238a01bf4d6d2ea6bae3","revisions":{"7c6c7e16ecab6e927e1f8442009fd2f99d293156":{"kind":"REWORK","_number":1,"created":"2022-05-12 20:39:32.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/36/841636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/36/841636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/841636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/841636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/841636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/36/841636/1"}}},"commit":{"parents":[{"commit":"936c2850f6f5d162af1d97317ab666673ebfdc79","subject":"rendering HTML for versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/936c2850f6f5d162af1d97317ab666673ebfdc79"}]}],"author":{"name":"Adam Young","email":"adam@younglogic.com","date":"2022-05-12 20:37:44.000000000","tz":-240},"committer":{"name":"Adam Young","email":"adam@younglogic.com","date":"2022-05-12 20:37:44.000000000","tz":-240},"subject":"Change Header to Basic auth on authenticaltion failure.","message":"Change Header to Basic auth on authenticaltion failure.\n\nChange-Id: I070d034926507f9b74c448c8cfa91f66ac568adf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7c6c7e16ecab6e927e1f8442009fd2f99d293156"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7c6c7e16ecab6e927e1f8442009fd2f99d293156"}]},"branch":"refs/heads/master"},"7c91198fc5731e927193fb20a0515563e1e1e9d9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-08-21 18:06:44.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/36/841636/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/36/841636/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/841636/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/841636/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/841636/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/36/841636/2"}}},"commit":{"parents":[{"commit":"2294884f0300aaaf6011a2342629528d355bb466","subject":"rendering HTML for versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2294884f0300aaaf6011a2342629528d355bb466"}]}],"author":{"name":"Adam Young","email":"adam@younglogic.com","date":"2024-08-21 18:06:38.000000000","tz":-240},"committer":{"name":"Adam Young","email":"adam@younglogic.com","date":"2024-08-21 18:06:38.000000000","tz":-240},"subject":"Change Header to Basic auth on authenticaltion failure.","message":"Change Header to Basic auth on authenticaltion failure.\n\nChange-Id: I070d034926507f9b74c448c8cfa91f66ac568adf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7c91198fc5731e927193fb20a0515563e1e1e9d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7c91198fc5731e927193fb20a0515563e1e1e9d9"}]},"branch":"refs/heads/master"},"0eef0759ba3c85e465a9238a01bf4d6d2ea6bae3":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-09-13 19:24:02.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/36/841636/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/36/841636/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/841636/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/841636/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/841636/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/36/841636/3"}}},"commit":{"parents":[{"commit":"2294884f0300aaaf6011a2342629528d355bb466","subject":"rendering HTML for versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2294884f0300aaaf6011a2342629528d355bb466"}]}],"author":{"name":"Adam Young","email":"adam@younglogic.com","date":"2024-08-21 18:06:38.000000000","tz":-240},"committer":{"name":"Adam Young","email":"adam@younglogic.com","date":"2024-09-13 19:24:02.000000000","tz":0},"subject":"Change Header to Basic auth on authentication failure.","message":"Change Header to Basic auth on authentication failure.\n\nChange-Id: I070d034926507f9b74c448c8cfa91f66ac568adf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0eef0759ba3c85e465a9238a01bf4d6d2ea6bae3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0eef0759ba3c85e465a9238a01bf4d6d2ea6bae3"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
