)]}'
{"id":"openstack%2Fkeystone~841638","triplet_id":"openstack%2Fkeystone~master~I05e150829c2ab3bb795d001e303413a084c38624","project":"openstack/keystone","branch":"master","topic":"html-rebase","attention_set":{"2218":{"account":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"last_update":"2022-05-12 22:03:25.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I05e150829c2ab3bb795d001e303413a084c38624","subject":"render html for RESTful Pages","status":"NEW","created":"2022-05-12 20:39:32.000000000","updated":"2024-09-13 20:20:45.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a46b79a52a14fd398a3e2b559ae17a9776f49062","_number":841638,"virtual_id_number":841638,"owner":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-09-13 20:13:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-12 21:19:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-12 22:03:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-21 19:16:01.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"CC"}],"messages":[{"id":"24d2d07b90dd91995e4c6bb69d4765f8e75f090c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2022-05-12 20:39:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c26cb0088dd043581875239a5b70574e6d3fb24","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 21:19:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f8810404e41498ba30ad8b10095720c : SUCCESS in 38m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d1dc84dbba54b9fa0aecb9d435aa2ef : SUCCESS in 36m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4b1d3164ac817e137c6290c08c3390f31ed7f388","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 22:03:25.000000000","message":"Patch Set 1: Verified-1\n\n(138 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f7226a1ab39423382614559ad9e5e5d : SUCCESS in 14m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c0251f7b095481f959398e3d6b6ca4b : FAILURE in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f1d3240949f431192ff2d79278f7e19 : SUCCESS in 12m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15916efc79e34718842845c7b7ba9211 : SUCCESS in 11m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5722be353aed477aab1f047b8269f2c1 : SUCCESS in 11m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c2065cd810449589af9da18ac7e253d : FAILURE in 7m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/58cf87812779415aae945af952202171 : SUCCESS in 55m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e5537c95b304acfbe838ad64d82653c : SUCCESS in 1h 18m 26s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/cfeb5d602fc24e6a95db9033fefbc32b : SUCCESS in 27m 46s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/602dde92fdb041939cfb53ceec48d394 : FAILURE in 17m 04s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3abf1fb7f5e42d899e7b4b0a354c2ac : SUCCESS in 27m 35s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/d9b51ab89a7a4e97a30dd0fee646f7c7 : SUCCESS in 28m 51s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/b885d635f5c3410aa444941e45eb44c7 : SUCCESS in 13m 33s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/474684c0379e4005b6cc0641881d4b9e : SUCCESS in 31m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6953e50ed2af48cfaa4c9427c15b5d9f : SUCCESS in 46m 38s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/9c147c3e629747fbb1372b8b4f1e97eb : SUCCESS in 46m 17s\n\nWarning:\n  Comments left for invalid file keystone/api/discovery.py\n  Comments left for invalid file keystone/server/flask/request_processing/middleware/auth_context.py","accounts_in_message":[],"_revision_number":1},{"id":"fbd3945684a8f30346de1096a06537ddd810e5e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2024-08-21 18:06:44.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a8993ff030a51f2f616a4e8f45459a191ff94177","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 19:03:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1358003affe4c53b830941b65a3e3dc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11ac04c7a789494cb40577baafa26495 : FAILURE in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1570b969b7c044f89c5633d7f2498975 : FAILURE in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d5862ce9dd94bc6b55cb66991d18d94 : FAILURE in 5m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/57cd90b7093d4a928d9f1369ecfff2a3 : FAILURE in 5m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c3b1c5dc256d40e18a72f186daefe328 : FAILURE in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7ddd8d66b5d48de9f412fdf23881adc : FAILURE in 7m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/c5c8bb5f6c9a4b3fb4cc8e78d5954658 : FAILURE in 50m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d42e5cc76ce044e892e9d4e646433048 : FAILURE in 22m 42s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/8e250937fea2443184cd28318ce80cb7 : FAILURE in 22m 40s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/0edaaa2727204f8bbfd8e0a377a4e5bc : FAILURE in 31m 50s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61b873e4e0f9406bb3116ee25c11403a : FAILURE in 24m 04s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/7c41b076cadb42a6bc90c959bdcc9258 : FAILURE in 26m 04s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/3d87e8499ac44f6ea3fe5d095c05f9ea : POST_FAILURE in 15m 56s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/88c95390aa3f4fd79d37a89deab8f60f : FAILURE in 20m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b801e30be48d436d95ee43030f09a39d : FAILURE in 21m 09s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/44b01f491d37415198f6255e36a01f85 : FAILURE in 23m 44s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/1fe1197f773943c7ae3f396a258ad033 : RETRY_LIMIT in 6m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c8b52f0f73f1086ffd93eec84be21502deedcab9","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-08-21 19:16:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"22bc9f4c4bfbc171446307e5b0888c1398d55612","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-08-21 19:16:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e921992d63d407d1ac7d86619238bc3aa6fe7e1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 21:44:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1276cfdac8c34b99bcbd6d58450d6851\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ae6e6431149c4116af02b97ebf0074d0 : FAILURE in 10m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/01b945d708a14bb4b7b5113f89f254d5 : FAILURE in 8m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f3136161b195b392dfe1548af6b5293ae0a78ea6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2024-09-13 19:26:49.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3c9a1116d099e20c94a09198d755a9d5ba4b92d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 20:13:10.000000000","message":"Patch Set 3: Verified-1\n\n(98 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed005cc0f7264e27a8e52db69f516172\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cd2f083c5a746068a8ebb1d8d5cbbe3 : FAILURE in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a0311672d1b45d88f3ecc5ce3f6f35c : FAILURE in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ecd2df74a6474d03836cc3831b0c326b : FAILURE in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d8e6124441934cec9a16b33087c4391c : FAILURE in 4m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/77e17245af014bc9915bf3d28101d641 : FAILURE in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc4e78ee720b4dc5a621d39dc6e31d26 : FAILURE in 4m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/97dc2a599f7148e68af283384c03f6e6 : FAILURE in 44m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9c03bc6c0fa43bebc839c984a1bd889 : FAILURE in 12m 49s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/71d1f409159d4462b0a4e63b308ef322 : FAILURE in 10m 46s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/c1ac0b67509b4c59855650775b17100c : FAILURE in 16m 56s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f99da2558bc4416a5ff947260c68993 : FAILURE in 23m 57s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/e62d1cb2659148e2a0d9fcfb07d69fbd : FAILURE in 15m 09s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/3b18855206a94f17afaa15d3eb4ecff6 : POST_FAILURE in 18m 25s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/1a19d458408d4df0beb6aad6d0d0935a : FAILURE in 15m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e8295aaf07f411cbef681f918d2b78c : FAILURE in 20m 58s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/4fffc2da70e44875905870e4b7f0c6c8 : FAILURE in 19m 37s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/f134b79ec7d54912ac2bbae600dc715e : POST_FAILURE in 5m 09s (non-voting)\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: file keystone/common/json2html.py has only 209 lines\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: file keystone/common/json2html.py has only 209 lines\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: file keystone/common/json2html.py has only 209 lines\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: file keystone/common/json2html.py has only 209 lines\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: file keystone/common/json2html.py has only 209 lines\u0027\n  Comments left for invalid file keystone/api/discovery.py\n  Comments left for invalid file keystone/server/flask/request_processing/middleware/auth_context.py","accounts_in_message":[],"_revision_number":3},{"id":"a46b79a52a14fd398a3e2b559ae17a9776f49062","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 20:20:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5728a1b38954be88e10ea0790bad9d6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/181c25879d844e88a36d52a204801d6f : FAILURE in 9m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b62f6e9681a247dc9919e3717d106fa2 : FAILURE in 9m 11s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"074bc6cdb1e2fcbeb8c24fd8d8a84fe875ac7765","revisions":{"196d3a3c9da943d0b22ae24b266fbae6c8b8294f":{"kind":"REWORK","_number":1,"created":"2022-05-12 20:39:32.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/38/841638/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/38/841638/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/841638/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/841638/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/841638/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/38/841638/1"}}},"commit":{"parents":[{"commit":"51ddd9993e04df38980f8563db357500f936a239","subject":"use basic auth to authenticate and fetch an unscoped token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/51ddd9993e04df38980f8563db357500f936a239"}]}],"author":{"name":"Adam Young","email":"adam@younglogic.com","date":"2022-05-12 20:39:21.000000000","tz":-240},"committer":{"name":"Adam Young","email":"adam@younglogic.com","date":"2022-05-12 20:39:21.000000000","tz":-240},"subject":"render html for RESTful Pages","message":"render html for RESTful Pages\n\nChange-Id: I05e150829c2ab3bb795d001e303413a084c38624\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/196d3a3c9da943d0b22ae24b266fbae6c8b8294f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/196d3a3c9da943d0b22ae24b266fbae6c8b8294f"}]},"branch":"refs/heads/master"},"4d604857621f30153a91f06256445cb29e9b0169":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-08-21 18:06:44.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/38/841638/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/38/841638/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/841638/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/841638/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/841638/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/38/841638/2"}}},"commit":{"parents":[{"commit":"9baab511f01d3377bebd0862acf49b8987a7716b","subject":"use basic auth to authenticate and fetch an unscoped token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9baab511f01d3377bebd0862acf49b8987a7716b"}]}],"author":{"name":"Adam Young","email":"adam@younglogic.com","date":"2024-08-21 18:06:38.000000000","tz":-240},"committer":{"name":"Adam Young","email":"adam@younglogic.com","date":"2024-08-21 18:06:38.000000000","tz":-240},"subject":"render html for RESTful Pages","message":"render html for RESTful Pages\n\nChange-Id: I05e150829c2ab3bb795d001e303413a084c38624\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4d604857621f30153a91f06256445cb29e9b0169"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4d604857621f30153a91f06256445cb29e9b0169"}]},"branch":"refs/heads/master"},"074bc6cdb1e2fcbeb8c24fd8d8a84fe875ac7765":{"kind":"NO_CHANGE","_number":3,"created":"2024-09-13 19:26:49.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/38/841638/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/38/841638/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/841638/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/841638/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/841638/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/38/841638/3"}}},"commit":{"parents":[{"commit":"ffaad701050e9e61586d9357b8601c283457dfc4","subject":"use basic auth to authenticate and fetch an unscoped token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ffaad701050e9e61586d9357b8601c283457dfc4"}]}],"author":{"name":"Adam Young","email":"adam@younglogic.com","date":"2024-08-21 18:06:38.000000000","tz":-240},"committer":{"name":"Adam Young","email":"adam@younglogic.com","date":"2024-09-13 19:26:49.000000000","tz":0},"subject":"render html for RESTful Pages","message":"render html for RESTful Pages\n\nChange-Id: I05e150829c2ab3bb795d001e303413a084c38624\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/074bc6cdb1e2fcbeb8c24fd8d8a84fe875ac7765"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/074bc6cdb1e2fcbeb8c24fd8d8a84fe875ac7765"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
