)]}'
{"id":"openstack%2Fkeystone~842259","triplet_id":"openstack%2Fkeystone~master~I1dea8ff64ea3fa0efed67f590c3b45bbdbe0f726","project":"openstack/keystone","branch":"master","topic":"lock-tox","attention_set":{},"removed_from_attention_set":{"2218":{"account":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"last_update":"2023-02-10 15:33:26.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1dea8ff64ea3fa0efed67f590c3b45bbdbe0f726","subject":"Lock down python in tox to 3.9, latest supported version","status":"ABANDONED","created":"2022-05-18 04:06:50.000000000","updated":"2023-02-10 15:33:26.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"322ca8de5b06b93055f632ba780ab2d11992d29a","_number":842259,"virtual_id_number":842259,"owner":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-18 05:47:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":-1,"date":"2022-07-25 15:08:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}],"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-18 04:43:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-18 05:47:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-27 15:24:08.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"CC"},{"updated":"2022-05-27 15:39:04.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"CC"},{"updated":"2022-07-25 15:08:16.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"3102c8f9fd3c5edb8b932f0e07121e5f8c42cab4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2022-05-18 04:06:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d640aa92159a8a35582b8cc6d6e90f42920c4cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 04:43:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/316abaddfe804ad1848f7fb37444d5c1 : SUCCESS in 30m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b5ad8b40c0f649ddb3a5cdc99bda3d6d : SUCCESS in 34m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf0003bda4b3fa97085a3f6398d147d1336b5344","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 05:47:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/853639f2c87d4d98a863ead1707bf75d : FAILURE in 3m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0decff10a0fc4440963148f28764cce7 : FAILURE in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8ecb9f79cc24637a4882374b6d3d4e0 : FAILURE in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efdd504a903145e2a75234b16f30fe5e : SUCCESS in 11m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e3cfdc20bd94846b1a71c2a01fe3896 : SUCCESS in 12m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2b57519c505473fb73e40f465e94689 : SUCCESS in 12m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa085364272a41c8a70a4f25d2bf7178 : FAILURE in 5m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/514cfa06969045dbacb14ce3d05a1d80 : SUCCESS in 59m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/59a7c71c5adc42818e5160246e56c68b : SUCCESS in 1h 35m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2cee75a546d4a34bc772946166ac480 : SUCCESS in 8m 32s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/5253f392ca9d4d9dbcf4b7ea855bd80d : SUCCESS in 36m 08s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/a9794e33ca56496e93d4400b76267d74 : FAILURE in 23m 42s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5b32a36a67b4c3eb079d4235c4b4eee : SUCCESS in 36m 40s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/b12da25f93fb4a3cba77fd748f7f1199 : SUCCESS in 43m 13s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/6ddfb518c98545cf870c30fee11bfec3 : SUCCESS in 21m 02s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/00d5fb1f345e4ef3a4d6f51cb6e5740e : SUCCESS in 39m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd42e849419f4184b5b4d77022500730 : SUCCESS in 58m 08s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/7c872bb0fa9e424d8931e0f17e71c864 : SUCCESS in 45m 38s","accounts_in_message":[],"_revision_number":1},{"id":"0bc45eaefacf669ec5e7cacb94f2f202df2c0799","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-05-27 15:24:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1115cc18e56bc187a1e810d57a68de5202fcb46a","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-05-27 15:39:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a7b0ee72400f5190d3cf9089bfe7afe7dd3667e","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-07-25 15:08:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"322ca8de5b06b93055f632ba780ab2d11992d29a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-02-10 15:33:26.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3326025313d4547a735435ac1832778cd276512d","revisions":{"3326025313d4547a735435ac1832778cd276512d":{"kind":"REWORK","_number":1,"created":"2022-05-18 04:06:50.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/59/842259/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/59/842259/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/59/842259/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/59/842259/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/59/842259/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/59/842259/1"}}},"commit":{"parents":[{"commit":"65e3e33b6b8be4d1a5b7c85456f3048244e4eb75","subject":"Merge \"typo fix in docstring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/65e3e33b6b8be4d1a5b7c85456f3048244e4eb75"}]}],"author":{"name":"Adam Young","email":"adam@younglogic.com","date":"2022-05-18 04:03:57.000000000","tz":-240},"committer":{"name":"Adam Young","email":"adam@younglogic.com","date":"2022-05-18 04:06:50.000000000","tz":-240},"subject":"Lock down python in tox to 3.9, latest supported version","message":"Lock down python in tox to 3.9, latest supported version\n\nWithout this change, pep8 fails on latest Fedora and Ubuntu\nwhich ship with Python 3.10 as the default.  PEP8 does\nnot yet run on Python 3.10.\n\nChange-Id: I1dea8ff64ea3fa0efed67f590c3b45bbdbe0f726\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3326025313d4547a735435ac1832778cd276512d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3326025313d4547a735435ac1832778cd276512d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
