)]}'
{"id":"openstack%2Fkeystone~842888","triplet_id":"openstack%2Fkeystone~stable%2Fxena~I5a2ba6087f63ba20489a32d1afb194def4b0c70e","project":"openstack/keystone","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"16465":{"account":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"last_update":"2022-07-20 20:21:13.000000000","reason":"Change was submitted"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2022-07-20 20:21:13.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-20 20:21:13.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2022-07-18 10:45:11.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}},"hashtags":[],"change_id":"I5a2ba6087f63ba20489a32d1afb194def4b0c70e","subject":"Xena-only: Fix wrong python job template used","status":"MERGED","created":"2022-05-22 14:25:44.000000000","updated":"2022-07-20 20:23:27.000000000","submitted":"2022-07-20 20:21:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842888","meta_rev_id":"cb3bb7d50453e887286a11714d333e5c6d6f2c6b","_number":842888,"virtual_id_number":842888,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-20 20:21:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2022-07-15 16:27:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2022-07-18 10:45:11.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2022-07-20 19:07:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-22 15:36:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-15 16:27:55.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2022-07-15 16:27:55.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2022-07-15 16:27:55.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2022-07-15 16:27:55.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"610aef2ce24f2d324699cf3f8a597058ce0791aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-22 14:25:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8d37e0299e70996cc6820570a70bffb10fb6981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-22 15:36:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3132fcdbf0249cf99673e92ec19d2b9 : SUCCESS in 15m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3960d2e4247446fa80c212018175501e : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7c913684257429d8d7cb84056833796 : SUCCESS in 12m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2cac5549bdbe4b05895b0199d3660802 : SUCCESS in 20m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a25a9a513c554048ba5fbb669f2b12c3 : SUCCESS in 17m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba5013c79e374b14b475db1b7133ad84 : SUCCESS in 10m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/911a61db5e73440cbfffec33bfb3cb17 : SUCCESS in 46m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5ebceaf5b884f24945b7b8c0409ad7f : SUCCESS in 1h 08m 47s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/edac54759acb441cb15d33b6e1054e85 : SUCCESS in 24m 52s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/83d50edc0b064a7e9a3a10ace89f98d9 : FAILURE in 28m 14s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/087b579cf5bc43c885f9c5f0d42073c9 : SUCCESS in 23m 41s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/caf6dc1aeabd46dd8e616e68638fe488 : SUCCESS in 24m 13s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/225923bd4afa4c3491bc875c04a28238 : SUCCESS in 18m 09s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/e35192802fef48c78ebebea16f3a0ab3 : FAILURE in 9m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd9969d507c04d8badf2526d816357e3 : SUCCESS in 49m 54s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/799e943d6d2e4eb594297bffb7c15922 : SUCCESS in 38m 26s","accounts_in_message":[],"_revision_number":1},{"id":"92fff8c65707c8f7bb18cc8bcd423f0b2c55a1d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-22 15:37:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c33e83a7f78426481aa9807717869a1 : SUCCESS in 1h 00m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/722cb5222efd40e393bef620c738f763 : SUCCESS in 1h 00m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2ee563f9b4c6f472dd069356594a7018b14422ab","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-07-15 16:27:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b1251630fcbe0c46ee4141d000b30f7d7a0e940b","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-07-18 10:45:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8144f8a0cbc3f7887eb70b2673eb312cd4c08b42","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-07-20 19:07:13.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"33b5a56405df2ed3be58b090fdddb3bb82692031","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 19:07:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2e6000555312f57849e5d1441cf008308c5777ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 20:21:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9354e574d1f44e0d91c3be56b7c68ec9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7dee8498f154fca8a3ca063c1c056ba : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/022d346d40cb45f9b6603f975cb942d5 : SUCCESS in 16m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/553c0656a07a439b97d0ed4bce866bfd : SUCCESS in 16m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5223a84bc442438ba1d989067a69fd3c : SUCCESS in 17m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/c408fe19fb9e419a8dba0a380c42ced7 : SUCCESS in 50m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/37b032f8a41c46f098d4f8b95b4be678 : SUCCESS in 1h 12m 34s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/105445d425fd4e5c80eb7247df909bad : SUCCESS in 27m 39s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/e5a68614bfea42d4bd77c1575ee18c68 : SUCCESS in 24m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/019c5951792e42b8860620dc30f74d4c : SUCCESS in 41m 33s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/28683d5f82e74c5bbc042770e595f066 : SUCCESS in 34m 01s","accounts_in_message":[],"_revision_number":1},{"id":"170787dbcead8c0d8fed49e0303d8e3425b69d7f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 20:21:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cb3bb7d50453e887286a11714d333e5c6d6f2c6b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 20:23:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96b579a1ed6e4edc929e1a9147034f79\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80de34f874624a4b91bac9e5eb76a435 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bd692531e98ce962c892059f32c2ba37ce3291e3","revisions":{"bd692531e98ce962c892059f32c2ba37ce3291e3":{"kind":"REWORK","_number":1,"created":"2022-05-22 14:25:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/842888/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/88/842888/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/842888/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/842888/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/842888/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/88/842888/1"}}},"commit":{"parents":[{"commit":"ca93f062df0af06616eefda3f2574ea814ee53bd","subject":"Fix issue with LDAP backend returning bytes instead of string","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ca93f062df0af06616eefda3f2574ea814ee53bd"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-22 14:25:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-22 14:25:20.000000000","tz":540},"subject":"Xena-only: Fix wrong python job template used","message":"Xena-only: Fix wrong python job template used\n\nWe should use the template corresponding to the release. This change\nreplaces the wrong template(victoria template) by the appropriate one\n(xena template).\n\nChange-Id: I5a2ba6087f63ba20489a32d1afb194def4b0c70e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bd692531e98ce962c892059f32c2ba37ce3291e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bd692531e98ce962c892059f32c2ba37ce3291e3"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[]}
