)]}'
{"id":"openstack%2Fkeystone~849724","triplet_id":"openstack%2Fkeystone~master~I539047bffa8b14f9d8d0adca8d215dd57745ec85","project":"openstack/keystone","branch":"master","topic":"project-tags","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2022-10-04 13:40:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I539047bffa8b14f9d8d0adca8d215dd57745ec85","subject":"Emit tags notification on project creation","status":"ABANDONED","created":"2022-07-13 17:57:24.000000000","updated":"2022-10-04 13:40:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5dcd2b0509906f2bf49775f36b21976ed39ebb47","_number":849724,"virtual_id_number":849724,"owner":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-13 18:45:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-13 19:46:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-26 15:47:36.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"371d63318514352140815e0b5dd926ec9276f718","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-07-13 17:57:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f370060b9bd4c443e24e2c2311c9563c2cc0356","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-07-13 17:58:07.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"8d4ef74c6600418170023368a11edcc75a77b491","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 18:45:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7084dd0f0d40460fb4488be85c13187c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae2f0b1e4fd548f9b87f0a56267eeb0b : SUCCESS in 45m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1908a613ea1348bcaa0a91fdcf86b59b : SUCCESS in 46m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0bf9d109ca30920ca95f390c41a7e4d8e5f8c3f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 19:46:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90967c3902fb4cc3897908178f062fb2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a185501270543f797202df1a87e1c4a : SUCCESS in 14m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51db081902144b238ab8bcf744abc1d2 : SUCCESS in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76d3131713e94ddf831e67ce92160634 : SUCCESS in 13m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75c65f27d38546b290059219e11b6f29 : SUCCESS in 13m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/691f95bf7955407190a31785c7e9a8b5 : SUCCESS in 13m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dfb5acc0a094a4390779ad7cdcf470f : SUCCESS in 15m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/ff7f440c6cc540f5aedde2a80532ca8c : SUCCESS in 1h 11m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97be3e9a92fb4bf79d9cd5063f1a91b4 : SUCCESS in 1h 44m 27s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d2a4040f0e2b4961b776525f55a818e9 : SUCCESS in 43m 35s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/c137901ee1bb494297c1058344d3a199 : SUCCESS in 56m 48s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50534c1c06f94d3e829855a5155452ff : SUCCESS in 42m 22s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/b9566f359cf04bc0bcec9664f4f9a2b4 : SUCCESS in 40m 55s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c05a970838c748e0aa098929aeb9117a : SUCCESS in 16m 21s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/431ab17c97934993afdd97ff08a5ca37 : SUCCESS in 40m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39867986d2714d7591435a5f496c75f3 : SUCCESS in 1h 03m 00s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/b0329f3b66ae40d3869d01e175f3f9c2 : SUCCESS in 39m 08s","accounts_in_message":[],"_revision_number":1},{"id":"b21032a99c2913778a7c2438206f4145173e1027","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-07-25 22:51:30.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ae7d8627035dc0696fc2d9bda2c8b81b3eca10b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 23:19:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e01fdae03b7a443f9a59f83dc75c6c0b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d93311ab1b54ed2ab75610f300a3399 : SUCCESS in 26m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/459d3c65f8fa49ca8994a60b8c5f728b : SUCCESS in 20m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dc437a9aa6d5a82c29be94c4aafb0e8bcd8af07c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 00:20:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5a1a9521de94f13a15ef04a266f72d8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/938790250c504e2db4d08f0432de2897 : SUCCESS in 17m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d5e63f7317741ca91860cf739ada71e : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8251485ce223421ca654f0d9d1ffd6c6 : SUCCESS in 16m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85d6fb3850c64607a48f2f1b3407c97c : SUCCESS in 14m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/124b55820d2d4d41bc56574a4aa6fff8 : SUCCESS in 16m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6200c077a65447c59d969c465d4b666d : SUCCESS in 11m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/cf8d2b361602475cb04568fab86afeab : SUCCESS in 54m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/51d87715c4a84936a7f37fcec9ad12e1 : SUCCESS in 1h 27m 18s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d405b1101fc04e5797f93d7946728811 : SUCCESS in 29m 10s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/89d311af4acf4e7dbbf58aad2e70a9bb : SUCCESS in 37m 33s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8b61b799ab7436e83b590e3519f26bc : SUCCESS in 25m 20s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/bc19b926b8cc45039d2b74aee4fb7a30 : SUCCESS in 27m 02s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/5f2ae6700a8d458fa7bef7d70352f594 : SUCCESS in 13m 58s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ecb80dd40ab14f90931f06d44dfc35b2 : SUCCESS in 30m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f522a1e609946cd9396289f2f9955e4 : SUCCESS in 47m 59s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/4c4f5061023945828c1a717820a7eb1e : SUCCESS in 36m 22s","accounts_in_message":[],"_revision_number":2},{"id":"c220006d4d7716423bf1f24efc54c0563c74f691","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-08-11 20:04:19.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"d49af6abdba31c6e5f1c8f2f9f008020ecf89cf9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 20:42:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1349207e3377498d93f18f4686185fad\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a06a61f22d5b4bba87e840349b098e7b : SUCCESS in 35m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4de4d6a5ce3146459d42e7fb102fd39b : SUCCESS in 22m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f51e91922218a8caccb99ae2dc8c42eeca1363c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 21:55:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e945aa189dd64c2cb975de079e370d63\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db53ae21ada34255a65290e1137db523 : SUCCESS in 15m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5346d3713e264339ab9cad11dcbb2ba6 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95001fcd01d841e48cf1ec90bab2eb3f : SUCCESS in 13m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7112d87659d94eca8e32d462fa0db698 : SUCCESS in 14m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81bf6b7d68084566b641e5ff114fd924 : SUCCESS in 14m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2eb042bbdf74a6f9492523d6e07d0a6 : SUCCESS in 11m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/d02e5610121a4b2abf0e412415c5765a : SUCCESS in 1h 01m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dccfc1eb4a58435e99b29ff5327cc59e : SUCCESS in 1h 50m 50s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/2255933d6d5b471d9d640fa2d90df207 : SUCCESS in 36m 58s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/e4586503e881460fbfd738e687e0e7b8 : FAILURE in 27m 31s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56d5224af6bf4e3d93f48b5a6da857f6 : SUCCESS in 34m 45s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/cbce9f9cc4724c9994d02c74c4f4de00 : SUCCESS in 42m 25s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/51c206dc8c914922a6f2cd40cfa4d1f2 : SUCCESS in 19m 32s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/e1d550639f0545aca800f7202eda5d9b : SUCCESS in 51m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e95c6f51a18b4e608a35e60176516723 : SUCCESS in 56m 34s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/60b413ff6fc8430da5250bb15612d6fa : SUCCESS in 52m 18s","accounts_in_message":[],"_revision_number":3},{"id":"9b243599959097f2855b0074e289a31cd4e65604","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-10-04 13:39:38.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"5dcd2b0509906f2bf49775f36b21976ed39ebb47","tag":"autogenerated:gerrit:abandon","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-10-04 13:40:19.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1d32325bcc37628aa133f1b61c2694574bad2974","revisions":{"3dfbc5d67a3a8fbe8ca68371a885d6baed55039a":{"kind":"REWORK","_number":1,"created":"2022-07-13 17:57:24.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/24/849724/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/849724/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/849724/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/849724/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/849724/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/849724/1"}}},"commit":{"parents":[{"commit":"d7b1d57cae738183f8d85413e942402a8a4efb31","subject":"Merge \"Move fips job to centos-9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d7b1d57cae738183f8d85413e942402a8a4efb31"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-07-13 17:54:34.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-07-13 17:54:34.000000000","tz":-300},"subject":"Emit tags notification on project creation","message":"Emit tags notification on project creation\n\nCurrently when a project is created with given tags, keystone does\nnot emit any notifications that any tags were created. This is a\nsmall gap in behavior where each project tag API emits a notification\nwhen project tags are modified in any way.\n\nThis change adds in a notification for project tags being updated for\na project if the project is created with project tags.\n\nChange-Id: I539047bffa8b14f9d8d0adca8d215dd57745ec85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3dfbc5d67a3a8fbe8ca68371a885d6baed55039a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3dfbc5d67a3a8fbe8ca68371a885d6baed55039a"}]},"branch":"refs/heads/master"},"f3b9a06f493fec8a3e9ffec25b5d54a6b035ee23":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-07-25 22:51:30.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/24/849724/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/849724/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/849724/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/849724/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/849724/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/849724/2"}}},"commit":{"parents":[{"commit":"aee576a7c69e9937079f697b228f03aedd24e970","subject":"Merge \"Change error response status code in master branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/aee576a7c69e9937079f697b228f03aedd24e970"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-07-13 17:54:34.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-07-25 22:51:30.000000000","tz":0},"subject":"Emit tags notification on project creation","message":"Emit tags notification on project creation\n\nCurrently when a project is created with given tags, keystone does\nnot emit any notifications that any tags were created. This is a\nsmall gap in behavior where each project tag API emits a notification\nwhen project tags are modified in any way.\n\nThis change adds in a notification for project tags being updated for\na project if the project is created with project tags.\n\nChange-Id: I539047bffa8b14f9d8d0adca8d215dd57745ec85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f3b9a06f493fec8a3e9ffec25b5d54a6b035ee23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f3b9a06f493fec8a3e9ffec25b5d54a6b035ee23"}]},"branch":"refs/heads/master","description":"Rebase"},"c59d7a4e7bc2985b01d76a4732a3c30be47bbc8d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-08-11 20:04:19.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/24/849724/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/849724/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/849724/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/849724/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/849724/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/849724/3"}}},"commit":{"parents":[{"commit":"1dd6993d7b9b647810e6f495b62c37627c6e8658","subject":"Merge \"docs: Update docs to reflect migration to Alembic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1dd6993d7b9b647810e6f495b62c37627c6e8658"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-07-13 17:54:34.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-08-11 20:04:19.000000000","tz":0},"subject":"Emit tags notification on project creation","message":"Emit tags notification on project creation\n\nCurrently when a project is created with given tags, keystone does\nnot emit any notifications that any tags were created. This is a\nsmall gap in behavior where each project tag API emits a notification\nwhen project tags are modified in any way.\n\nThis change adds in a notification for project tags being updated for\na project if the project is created with project tags.\n\nChange-Id: I539047bffa8b14f9d8d0adca8d215dd57745ec85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c59d7a4e7bc2985b01d76a4732a3c30be47bbc8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c59d7a4e7bc2985b01d76a4732a3c30be47bbc8d"}]},"branch":"refs/heads/master","description":"Rebase"},"1d32325bcc37628aa133f1b61c2694574bad2974":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-10-04 13:39:38.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/24/849724/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/849724/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/849724/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/849724/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/849724/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/849724/4"}}},"commit":{"parents":[{"commit":"02db9263961bb1a630014536096bcfc72009dd65","subject":"Merge \"Update master for stable/zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/02db9263961bb1a630014536096bcfc72009dd65"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-07-13 17:54:34.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-10-04 13:39:38.000000000","tz":0},"subject":"Emit tags notification on project creation","message":"Emit tags notification on project creation\n\nCurrently when a project is created with given tags, keystone does\nnot emit any notifications that any tags were created. This is a\nsmall gap in behavior where each project tag API emits a notification\nwhen project tags are modified in any way.\n\nThis change adds in a notification for project tags being updated for\na project if the project is created with project tags.\n\nChange-Id: I539047bffa8b14f9d8d0adca8d215dd57745ec85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1d32325bcc37628aa133f1b61c2694574bad2974"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1d32325bcc37628aa133f1b61c2694574bad2974"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
