)]}'
{"id":"openstack%2Fkeystone~851051","triplet_id":"openstack%2Fkeystone~stable%2Fyoga~I4bb54aac9b950d59082a4468203a3249790839d7","project":"openstack/keystone","branch":"stable/yoga","topic":"bug/1926483-stable/yoga","attention_set":{},"removed_from_attention_set":{"16465":{"account":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"last_update":"2022-10-26 16:58:00.000000000","reason":"Change was submitted"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2022-10-26 16:58:00.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2022-10-26 15:14:57.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}},"hashtags":[],"change_id":"I4bb54aac9b950d59082a4468203a3249790839d7","subject":"Only log warnings about token length when length exceeds max_token_size","status":"MERGED","created":"2022-07-26 13:58:53.000000000","updated":"2022-10-27 06:18:45.000000000","submitted":"2022-10-26 16:58:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851051-bug/1926483-stable/yoga","cherry_pick_of_change":789417,"cherry_pick_of_patch_set":2,"meta_rev_id":"d72b0542acccef6f3ab16379965b4af71a24d6cd","_number":851051,"virtual_id_number":851051,"owner":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-26 16:58:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2022-08-19 15:09:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2022-10-26 15:14:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-27 06:18:45.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2022-10-26 15:14:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}],"reviewers":{"CC":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"}],"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-26 13:58:53.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"CC"},{"updated":"2022-07-26 15:21:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-26 15:43:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-19 15:09:28.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2022-08-19 15:09:44.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2022-08-19 15:09:44.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2022-10-27 06:18:45.000000000","updated_by":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"reviewer":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"state":"REVIEWER"}],"messages":[{"id":"8589bcd40436d55baff9e7ef06e35b75c488039b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-07-26 13:58:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b0853a54db8a35e2fe17f8ddf26a61f5e2365592","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 15:21:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20048d9390644343ae1578663dab8e4e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fac402ef8c094bd6baf6921de7ff9b27 : TIMED_OUT in 1h 01m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5633686c763a4819884f5e286a7f0fc5 : RETRY_LIMIT in 3m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0f7e3aa544ac6d7665fe927c8560bba702243c1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 15:43:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec361ac6dd2a46bca59922a9e60eed22\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/172e3e75ee7f4deaa31559c9154f68e6 : SUCCESS in 15m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6110fc6156064d7bb905bb677a389e05 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30e25ba64e054024ad4f1c262f52c0ae : SUCCESS in 13m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f30907c8098b481891be2fefe137fbc0 : SUCCESS in 13m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e742d12610844938256d4032866ef04 : SUCCESS in 13m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/e0c89b9df1d04b8f9fd5143aed094517 : SUCCESS in 58m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/578207e5aa884787aa816f3021637006 : SUCCESS in 1h 42m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3778100a87194ce19be9ee2501182ff5 : SUCCESS in 8m 23s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/3239d3451a154954a33c065169bdc296 : SUCCESS in 32m 28s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/123d2bd7e01f45e98edf16b5817910db : RETRY_LIMIT in 14m 27s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e311919d193e42fd83b8a1b30be2259d : SUCCESS in 32m 21s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/ba7356e5a72649ffb62b71cfddfadcb4 : SUCCESS in 32m 56s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/a63f1b14ad644ed5aede7fcf4e83a531 : SUCCESS in 15m 51s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/eb253d470ff04aa7998097e8005c4817 : SUCCESS in 37m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/184061d97a1d453cb748c426f910d98a : SUCCESS in 48m 40s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/98cdf125a06e4354bd456875172b5b6f : SUCCESS in 42m 37s","accounts_in_message":[],"_revision_number":1},{"id":"79fe937c28db6ea826d59ba9f53472ffd04ecf26","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-08-19 15:09:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"104407a8ac58bd4687abc6ffbb6f03388e5f4dd3","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-10-26 15:14:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"70938bcac7ffabb26fcc6b4eb25f8d0d1d55c1af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 15:15:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af4e838e8e4744a4bca6107d97dc28ed84cfe5ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 16:58:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e94753864a647fb8ad8b40eb1817ab5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed732534722f46ab99270082dd69d9dc : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b955c39fc0b4289909b9f979e29a3a5 : SUCCESS in 10m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb38a17627994e44bfbc6e6d6b9f5fc8 : SUCCESS in 12m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36b7f8f717eb4f2f81188d14acb5e2d7 : SUCCESS in 11m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/fd735224448b4b7b934483bbfdd68c9b : SUCCESS in 46m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/16bc9fb8bd1344f3a21532ac9c7e8708 : SUCCESS in 1h 37m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/019495a419bc44d3a8fb70689773175a : SUCCESS in 6m 57s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/5c4ba65dec2b4692ab32f98621eb6290 : SUCCESS in 40m 26s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/8f1d326352274c0cbb1841105b53a76f : SUCCESS in 26m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24a43dd952a5420a80e8ec62dd38abc4 : SUCCESS in 1h 03m 58s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/e7bf863e574a40be88531f8a92251aa2 : SUCCESS in 43m 16s","accounts_in_message":[],"_revision_number":1},{"id":"4bf00c9d1557a564a66530fe8c3ba72a5b91ed85","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 16:58:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2a2616236849e3112d392841f62120e9a1f05601","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 16:59:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4a4d4e6ac06446a8f74aaead9660df3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f555ac7459ac419d80d8fc84be00b8fa : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a583c80f4be74794b0635530b45818f4 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1},{"id":"d72b0542acccef6f3ab16379965b4af71a24d6cd","author":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"date":"2022-10-27 06:18:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aaff84323b020b88682995cf54f4497fe9182815","revisions":{"aaff84323b020b88682995cf54f4497fe9182815":{"kind":"REWORK","_number":1,"created":"2022-07-26 13:58:53.000000000","uploader":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"ref":"refs/changes/51/851051/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/51/851051/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/851051/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/851051/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/51/851051/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/51/851051/1"}}},"commit":{"parents":[{"commit":"74be2e2465277aa93932fb9cabb3781451fcbe8a","subject":"Merge \"Yoga-only: Fix wrong python job template used\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/74be2e2465277aa93932fb9cabb3781451fcbe8a"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-05-03 20:37:35.000000000","tz":0},"committer":{"name":"David Wilde","email":"dwilde@redhat.com","date":"2022-07-26 13:58:53.000000000","tz":0},"subject":"Only log warnings about token length when length exceeds max_token_size","message":"Only log warnings about token length when length exceeds max_token_size\n\nPreviously, the fernet token provider would log warnings when a fernet\ntoken exceeded 255 characters, which is common for LDAP-backed\ndeployments. The warning is always issued, even when operators configure\nkeystone\u0027s max_token_size to a higher value, causing confusion because\nit appears the configuration value is silently ignored.\n\nThis commit fixes that issue by using the max_token_size configuration\nparameter consistently in the fernet token provider.\n\nCloses-Bug: 1926483\n\nChange-Id: I4bb54aac9b950d59082a4468203a3249790839d7\n(cherry picked from commit 68bfb685d12937dde11d1a335bd992203ec7c293)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/aaff84323b020b88682995cf54f4497fe9182815"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/aaff84323b020b88682995cf54f4497fe9182815"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[]}
