)]}'
{"id":"openstack%2Fkeystone~852516","triplet_id":"openstack%2Fkeystone~master~I86e0fefabb43800a7d667821c3aaebe3e2a2d03c","project":"openstack/keystone","branch":"master","hashtags":[],"change_id":"I86e0fefabb43800a7d667821c3aaebe3e2a2d03c","subject":"Remove unicode literal from code","status":"ABANDONED","created":"2022-08-09 06:37:16.000000000","updated":"2022-08-09 10:47:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"07335e215804becb42580bbfbb1ba35a2f29f418","_number":852516,"virtual_id_number":852516,"owner":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-08-09 06:50:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2022-08-09 06:50:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"46d6c07011c30d61c2da43c90dfab3b5247e1cfe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"date":"2022-08-09 06:37:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"250c7eba7d8af4e791cd3e56af92a241eb1a9af1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 06:50:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d2e11d8f812434bb165ae300ffa7514\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d469e1bd02d04cb7a5247305f3d4d2bb : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ea8043316784048af3c65b0283310ae : SUCCESS in 11m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d86c848a19a84f189fdbca52a7f4f6ad : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":1},{"id":"07335e215804becb42580bbfbb1ba35a2f29f418","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"date":"2022-08-09 10:47:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56ee071e6d0b7fe4c9a2c4d259567871beb88b04","revisions":{"56ee071e6d0b7fe4c9a2c4d259567871beb88b04":{"kind":"REWORK","_number":1,"created":"2022-08-09 06:37:16.000000000","uploader":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"ref":"refs/changes/16/852516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/16/852516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/16/852516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/16/852516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/16/852516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/16/852516/1"}}},"commit":{"parents":[{"commit":"1dd6993d7b9b647810e6f495b62c37627c6e8658","subject":"Merge \"docs: Update docs to reflect migration to Alembic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1dd6993d7b9b647810e6f495b62c37627c6e8658"}]}],"author":{"name":"chenwei","email":"chen.wei@99cloud.net","date":"2022-08-09 06:34:46.000000000","tz":480},"committer":{"name":"chenwei","email":"chen.wei@99cloud.net","date":"2022-08-09 06:37:00.000000000","tz":480},"subject":"Remove unicode literal from code","message":"Remove unicode literal from code\n\nAll strings are considered as unicode literal string from Python 3.\n\nCloses bug: #1984040\n\nChange-Id: I86e0fefabb43800a7d667821c3aaebe3e2a2d03c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/56ee071e6d0b7fe4c9a2c4d259567871beb88b04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/56ee071e6d0b7fe4c9a2c4d259567871beb88b04"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
