)]}'
{"id":"openstack%2Fkeystone~860928","triplet_id":"openstack%2Fkeystone~master~Ib1c391b06b167c55435360b81bb67c19a63c8b26","project":"openstack/keystone","branch":"master","topic":"bp/support-oauth2-mtls","attention_set":{},"removed_from_attention_set":{"16465":{"account":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"last_update":"2025-01-17 14:00:07.000000000","reason":"Change was submitted"},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2025-01-17 14:00:07.000000000","reason":"Change was submitted"},"5046":{"account":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"last_update":"2025-01-17 14:00:07.000000000","reason":"Change was submitted"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2025-01-15 15:09:46.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"34712":{"account":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"last_update":"2025-01-17 14:00:07.000000000","reason":"Change was submitted"},"34634":{"account":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"last_update":"2025-01-17 14:00:07.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2025-01-17 12:30:35.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},"33455":{"account":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"last_update":"2025-01-17 14:00:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib1c391b06b167c55435360b81bb67c19a63c8b26","subject":"Add doc of OAuth 2.0 Mutual-TLS Authenticate","status":"MERGED","created":"2022-10-11 08:46:56.000000000","updated":"2025-01-17 14:02:36.000000000","submitted":"2025-01-17 14:00:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"860928-bp/support-oauth2-mtls","meta_rev_id":"1eeaaf2defa9228f162bc87abace91be35d825ef","_number":860928,"virtual_id_number":860928,"owner":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-17 14:00:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2025-01-15 15:09:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2025-01-17 12:30:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2025-01-17 12:30:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"}],"reviewers":{"CC":[{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"}],"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-11 09:09:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-24 02:02:54.000000000","updated_by":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"reviewer":{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"},"state":"CC"},{"updated":"2022-11-04 10:47:06.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"state":"REVIEWER"},{"updated":"2022-11-04 10:51:05.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2022-11-04 10:51:05.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2022-11-04 10:51:05.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2022-11-04 10:51:05.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2022-11-04 10:51:05.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2022-11-04 10:52:49.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"state":"REVIEWER"}],"messages":[{"id":"deecafde914d09c49373b2bc2841e274fb9ed358","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"date":"2022-10-11 08:46:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44d785935e2fae71ff62c66d3134728babfb5cda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"date":"2022-10-11 08:48:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2cfce3ebd68454638a547e4ff4ea956a33127ad2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 09:09:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14f3f4ef3bf24830b5546bb0d395b0eb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6409b96879734546861c2c7ebfde1d47 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e93ef40cf51942b2ad3d35fe665dc1d7 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4071627451c5416ebbb76f3d36b8eb4c : SUCCESS in 14m 16s","accounts_in_message":[],"_revision_number":2},{"id":"f892f5addda2fad5614ba9acfba48310625a1d42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"},"date":"2022-10-11 11:35:36.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ed2984e0e7479095ea9e5a777af523abb33fac98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 11:53:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0034f14e29d4f299005908bdb7e8b4f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c66f97a2c65c4ffc9bbd13d43328d3a9 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64307e4996e64d98bf17538a54c7e00c : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e393469f73a346799aaea2b8d90738f6 : SUCCESS in 12m 53s","accounts_in_message":[],"_revision_number":3},{"id":"0d3b23ca7642e9110c2d09a83a78d7cb9e243dc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"date":"2022-10-24 02:02:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4312ce5f02383a7a97a28e602a9275ac359cc517","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"date":"2022-10-24 02:19:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"abc1d4d1f9ebe1ba8ecf66cf2a1648c74e7f81b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 02:37:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7cbab0999234f27b7a335d0208ab2a8\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c94d464ac2e84db9a391406f29668b3e : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e60e47853a244d628aba40b8cdc16452 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37eaf13568b0455a9c77190cd2cbd1c2 : SUCCESS in 11m 31s","accounts_in_message":[],"_revision_number":5},{"id":"cd863517ff4868408cdfabb3e72f52b574ad954a","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-12-09 14:43:32.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1744e2de390fbc7c55307520bfbfb8a08e8edf6c","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2023-03-02 11:11:58.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"90a93d89a03a9c699114914416c1a07ebee63c7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"date":"2023-07-21 01:29:51.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"0645330d59fb82678e511cf13df15886efb5f959","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-21 01:40:10.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dd65c5c9efa4055993873b8f2507d73\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9282d0bf1ece440f806a3a10eee76bbc : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3228dda85a2d44db8384f94cd69feef2 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf4d78f861d7410bbba9e316ea50d687 : SUCCESS in 9m 29s","accounts_in_message":[],"_revision_number":6},{"id":"197922dcd97145bbec723eb47a9a725db0cbd58f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-09-11 14:38:32.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"8dc9ca808e37a08bddfec60df064343d4cff9ccf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 14:54:50.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f763b0014f04762960ba3c7cdbca53e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ece5a59d44cb41edb0a11ed2ec96256f : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b1168b61211491c89b3b55a84cc7c0a : FAILURE in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12c33f966d434ef480c53dc65c8ca2a5 : SUCCESS in 11m 32s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/61a360a6633d475bae13c07c27b42178 : SUCCESS in 6m 27s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"54015d6452ce95f4a679ec454eece581b8436e34","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-09-12 12:59:34.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"01a88fa09a0a3b4e22c2f753a3d7815d3546b3d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 13:16:32.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/463888e93c0540d3893ff7b83c23c78d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ad6ba0ce0f184cb18688ad641368fd97 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1d872526e374cb99d7ae88bd7372807 : FAILURE in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81c857d1a12242398c4b1f776ff51133 : SUCCESS in 11m 07s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/88427aa50a30453aa7c7e5302bbc6dbf : SUCCESS in 5m 17s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"53414ccd2587f1abcd4f7eb60bcfb94774f422df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2024-10-01 08:38:48.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"3476516ae4d50e78d711a32584561f42eaabd163","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 08:50:37.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e810195e66549f28abc4d9058826a19\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8f34cc5c52e245839a0d8484ca98b17e : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d76d62159674df297afd4191c323bec : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5bc500be5f448feb921ddb01141dab8 : SUCCESS in 11m 02s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/5694f36e378f4efcbb2b15b70d886c9f : SUCCESS in 7m 08s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f02d0829aa2e4e5b6d1eee50a46db383e16c18d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2024-12-25 07:17:49.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"62af1dab4f8c0871fd7e2a9dcfda9cbe6470a64c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-25 08:00:16.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf733f5c914144ae809dce17493a7916\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3cafa138d0ea4ba894098863e0b36b13 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a994bb55485f45e49c918de0eafce874 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61c311369662466f8343a208357bf394 : SUCCESS in 12m 03s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/36785d960706448abebde2f72ed91a3f : SUCCESS in 6m 01s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"0909b0a1f6bebfd651c4c283b93f611406e16f3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2024-12-26 05:31:44.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"a061a549c05449eb933dc5f37ed1f9317cca20d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-26 05:51:39.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d374e12b12943098477f3525ab788bb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/82252c567a40403fa5489ae18524885d : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6801984d00d48108bfb180867984035 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bd1a036a7744ad4b04fe84e4bdffb84 : SUCCESS in 12m 07s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/c457fe9464264f3b84ed8ff7c521f424 : SUCCESS in 6m 02s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d27d3eaba36dbcd2e61795cc95c385f6a65cef8c","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2024-12-26 06:26:22.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"025ee4793136b3c5cd992901b516f2c0688921e1","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2025-01-15 15:09:46.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"b14ddb3fa34c79da667b9a5a5dbddc069b16e7ee","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-01-17 12:30:35.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"e54eef4e595ead8a1f17540acd05a02364378148","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 12:31:01.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"c550576f3afb4c4d1b825a868f4298ccf0b6e322","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 14:00:07.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32c089fe9ab84c479d75b491db9058dd\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/11fb59aa77fa40a6ba8bca5663599c89 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1860340825742229ac481adfa4e8a7d : SUCCESS in 8m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32215c2817ee4e238418887644ad2ffb : SUCCESS in 7m 46s","accounts_in_message":[],"_revision_number":10},{"id":"2743aa99a395916c43b27e766fc31aaa939f8747","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 14:00:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"1eeaaf2defa9228f162bc87abace91be35d825ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 14:02:36.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0cd834a2cc2435b9adc87086711a4a7\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/40d1e24316044550827093452617c672 : SUCCESS in 36s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa74624b34114184b6270b5b91e3151b : SUCCESS in 47s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"d16cb8771d30fbceb377463dc54b74db84ab709e","revisions":{"e625d1d3ab0c8e484ff54b7cd2437fb1ec40d50d":{"kind":"REWORK","_number":1,"created":"2022-10-11 08:46:56.000000000","uploader":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"ref":"refs/changes/28/860928/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/28/860928/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/28/860928/1"}}},"commit":{"parents":[{"commit":"3fec3b863e144573dc4d9c17b541cf606d271561","subject":"Add doc of OAuth2.0 Client Credentials Grant Flow","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3fec3b863e144573dc4d9c17b541cf606d271561"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-10-11 08:40:15.000000000","tz":540},"committer":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-10-11 08:48:53.000000000","tz":540},"subject":"Add doc of OAuth 2.0 Mutual-TLS Authenticate","message":"Add doc of OAuth 2.0 Mutual-TLS Authenticate\n\nThe OAuth2.0 Access Token API is modified, support to get an OAuth2.0\ncertificate-bound access token from the keystone identity server with\nOAuth 2.0 credentials and Mutual-TLS certificates.\n\nImplements: blueprint support-oauth2-mtls\nChange-Id: Ib1c391b06b167c55435360b81bb67c19a63c8b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e625d1d3ab0c8e484ff54b7cd2437fb1ec40d50d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e625d1d3ab0c8e484ff54b7cd2437fb1ec40d50d"}]},"branch":"refs/heads/master"},"84be921a8bd3f412a39cf5c130db01b8a5f19fa2":{"kind":"REWORK","_number":2,"created":"2022-10-11 08:48:43.000000000","uploader":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"ref":"refs/changes/28/860928/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/28/860928/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/28/860928/2"}}},"commit":{"parents":[{"commit":"3fec3b863e144573dc4d9c17b541cf606d271561","subject":"Add doc of OAuth2.0 Client Credentials Grant Flow","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3fec3b863e144573dc4d9c17b541cf606d271561"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-10-11 08:40:15.000000000","tz":540},"committer":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-10-11 08:50:41.000000000","tz":540},"subject":"Add doc of OAuth 2.0 Mutual-TLS Authenticate","message":"Add doc of OAuth 2.0 Mutual-TLS Authenticate\n\nThe OAuth2.0 Access Token API is modified, support to get an OAuth2.0\ncertificate-bound access token from the keystone identity server with\nOAuth 2.0 credentials and Mutual-TLS certificates.\n\nImplements: blueprint support-oauth2-mtls\nChange-Id: Ib1c391b06b167c55435360b81bb67c19a63c8b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/84be921a8bd3f412a39cf5c130db01b8a5f19fa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/84be921a8bd3f412a39cf5c130db01b8a5f19fa2"}]},"branch":"refs/heads/master"},"e211024e5ed82da038bc99099ee0286a2a0e36b3":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-10-11 11:35:36.000000000","uploader":{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"},"ref":"refs/changes/28/860928/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/28/860928/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/28/860928/3"}}},"commit":{"parents":[{"commit":"3fec3b863e144573dc4d9c17b541cf606d271561","subject":"Add doc of OAuth2.0 Client Credentials Grant Flow","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3fec3b863e144573dc4d9c17b541cf606d271561"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-10-11 08:40:15.000000000","tz":540},"committer":{"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","date":"2022-10-11 11:35:36.000000000","tz":0},"subject":"[WIP]Add doc of OAuth 2.0 Mutual-TLS Authenticate","message":"[WIP]Add doc of OAuth 2.0 Mutual-TLS Authenticate\n\nThe OAuth2.0 Access Token API is modified, support to get an OAuth2.0\ncertificate-bound access token from the keystone identity server with\nOAuth 2.0 credentials and Mutual-TLS certificates.\n\nImplements: blueprint support-oauth2-mtls\nChange-Id: Ib1c391b06b167c55435360b81bb67c19a63c8b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e211024e5ed82da038bc99099ee0286a2a0e36b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e211024e5ed82da038bc99099ee0286a2a0e36b3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3856d58483dd420345f1cdd1e562e975d687200f":{"kind":"REWORK","_number":4,"created":"2022-10-24 02:02:54.000000000","uploader":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"ref":"refs/changes/28/860928/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/28/860928/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/28/860928/4"}}},"commit":{"parents":[{"commit":"c141e6164e77f5ebb1369016d06de23a308ef510","subject":"trivial: Fix typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c141e6164e77f5ebb1369016d06de23a308ef510"}]}],"author":{"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","date":"2022-04-15 09:56:20.000000000","tz":0},"committer":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-10-24 02:05:39.000000000","tz":540},"subject":"Add doc of OAuth 2.0 Mutual-TLS Authenticate","message":"Add doc of OAuth 2.0 Mutual-TLS Authenticate\n\nThe OAuth2.0 Access Token API is modified, support to get an OAuth2.0\ncertificate-bound access token from the keystone identity server with\nOAuth 2.0 credentials and Mutual-TLS certificates.\n\nImplements: blueprint support-oauth2-mtls\nChange-Id: Ib1c391b06b167c55435360b81bb67c19a63c8b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3856d58483dd420345f1cdd1e562e975d687200f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3856d58483dd420345f1cdd1e562e975d687200f"}]},"branch":"refs/heads/master"},"a9a6c68d8c61c4c56c2472c52cd719e937cc136b":{"kind":"REWORK","_number":5,"created":"2022-10-24 02:19:13.000000000","uploader":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"ref":"refs/changes/28/860928/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/28/860928/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/28/860928/5"}}},"commit":{"parents":[{"commit":"3fec3b863e144573dc4d9c17b541cf606d271561","subject":"Add doc of OAuth2.0 Client Credentials Grant Flow","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3fec3b863e144573dc4d9c17b541cf606d271561"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-10-24 02:18:51.000000000","tz":540},"committer":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-10-24 02:20:55.000000000","tz":540},"subject":"Add doc of OAuth 2.0 Mutual-TLS Authenticate","message":"Add doc of OAuth 2.0 Mutual-TLS Authenticate\n\nThe OAuth2.0 Access Token API is modified, support to get an OAuth2.0\ncertificate-bound access token from the keystone identity server with\nOAuth 2.0 credentials and Mutual-TLS certificates.\n\nImplements: blueprint support-oauth2-mtls\nChange-Id: Ib1c391b06b167c55435360b81bb67c19a63c8b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a9a6c68d8c61c4c56c2472c52cd719e937cc136b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a9a6c68d8c61c4c56c2472c52cd719e937cc136b"}]},"branch":"refs/heads/master"},"515c3e2285f86a0049bd1ba17d7734eb7e1f2f8d":{"kind":"REWORK","_number":6,"created":"2023-07-21 01:29:51.000000000","uploader":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"ref":"refs/changes/28/860928/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/28/860928/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/28/860928/6"}}},"commit":{"parents":[{"commit":"74708e6124821f6941209668633ef7ca9e4981c9","subject":"Merge \"doc: Correct typo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/74708e6124821f6941209668633ef7ca9e4981c9"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-10-24 02:18:51.000000000","tz":540},"committer":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2023-07-21 01:31:45.000000000","tz":540},"subject":"Add doc of OAuth 2.0 Mutual-TLS Authenticate","message":"Add doc of OAuth 2.0 Mutual-TLS Authenticate\n\nThe OAuth2.0 Access Token API is modified, support to get an OAuth2.0\ncertificate-bound access token from the keystone identity server with\nOAuth 2.0 credentials and Mutual-TLS certificates.\n\nImplements: blueprint support-oauth2-mtls\nChange-Id: Ib1c391b06b167c55435360b81bb67c19a63c8b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/515c3e2285f86a0049bd1ba17d7734eb7e1f2f8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/515c3e2285f86a0049bd1ba17d7734eb7e1f2f8d"}]},"branch":"refs/heads/master"},"c504f4bb4f691f437e445caec84c1f41525111ed":{"kind":"REWORK","_number":7,"created":"2024-09-11 14:38:32.000000000","uploader":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"ref":"refs/changes/28/860928/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/28/860928/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/28/860928/7"}}},"commit":{"parents":[{"commit":"97431ec99b255f3e04d4b21d38412e5d6b798618","subject":"Merge \"Remove logic to support pysaml2\u003c3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/97431ec99b255f3e04d4b21d38412e5d6b798618"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-10-24 02:18:51.000000000","tz":540},"committer":{"name":"Dave Wilde","email":"dwilde@redhat.com","date":"2024-09-11 14:38:16.000000000","tz":-300},"subject":"Add doc of OAuth 2.0 Mutual-TLS Authenticate","message":"Add doc of OAuth 2.0 Mutual-TLS Authenticate\n\nThe OAuth2.0 Access Token API is modified, support to get an OAuth2.0\ncertificate-bound access token from the keystone identity server with\nOAuth 2.0 credentials and Mutual-TLS certificates.\n\nImplements: blueprint support-oauth2-mtls\nChange-Id: Ib1c391b06b167c55435360b81bb67c19a63c8b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c504f4bb4f691f437e445caec84c1f41525111ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c504f4bb4f691f437e445caec84c1f41525111ed"}]},"branch":"refs/heads/master"},"4eaa5a862cd6e363c4ecfa00fb4993cc701f454f":{"kind":"REWORK","_number":8,"created":"2024-10-01 08:38:48.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/28/860928/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/28/860928/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/28/860928/8"}}},"commit":{"parents":[{"commit":"9acac37b2b2005972ca2a9cbec3aae53bc419300","subject":"Merge \"Prevent circlular imports in keystone.api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9acac37b2b2005972ca2a9cbec3aae53bc419300"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-10-24 02:18:51.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","date":"2024-10-01 08:05:04.000000000","tz":540},"subject":"Add doc of OAuth 2.0 Mutual-TLS Authenticate","message":"Add doc of OAuth 2.0 Mutual-TLS Authenticate\n\nThe OAuth2.0 Access Token API is modified, support to get an OAuth2.0\ncertificate-bound access token from the keystone identity server with\nOAuth 2.0 credentials and Mutual-TLS certificates.\n\nImplements: blueprint support-oauth2-mtls\nChange-Id: Ib1c391b06b167c55435360b81bb67c19a63c8b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4eaa5a862cd6e363c4ecfa00fb4993cc701f454f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4eaa5a862cd6e363c4ecfa00fb4993cc701f454f"}]},"branch":"refs/heads/master"},"b0106bd0716f952e2bd0318034347500ffdd7963":{"kind":"REWORK","_number":9,"created":"2024-12-25 07:17:49.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/28/860928/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/28/860928/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/28/860928/9"}}},"commit":{"parents":[{"commit":"6bcc050ddd62da5860c1969d6d17e0e8aed76bd8","subject":"Merge \"Add JSON schema and validation for `implied role`\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6bcc050ddd62da5860c1969d6d17e0e8aed76bd8"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-10-24 02:18:51.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","date":"2024-12-24 18:13:25.000000000","tz":540},"subject":"Add doc of OAuth 2.0 Mutual-TLS Authenticate","message":"Add doc of OAuth 2.0 Mutual-TLS Authenticate\n\nThe OAuth2.0 Access Token API is modified, support to get an OAuth2.0\ncertificate-bound access token from the keystone identity server with\nOAuth 2.0 credentials and Mutual-TLS certificates.\n\nImplements: blueprint support-oauth2-mtls\nChange-Id: Ib1c391b06b167c55435360b81bb67c19a63c8b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b0106bd0716f952e2bd0318034347500ffdd7963"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b0106bd0716f952e2bd0318034347500ffdd7963"}]},"branch":"refs/heads/master"},"d16cb8771d30fbceb377463dc54b74db84ab709e":{"kind":"REWORK","_number":10,"created":"2024-12-26 05:31:44.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/28/860928/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/28/860928/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/860928/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/28/860928/10"}}},"commit":{"parents":[{"commit":"6bcc050ddd62da5860c1969d6d17e0e8aed76bd8","subject":"Merge \"Add JSON schema and validation for `implied role`\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6bcc050ddd62da5860c1969d6d17e0e8aed76bd8"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-10-24 02:18:51.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","date":"2024-12-26 05:31:21.000000000","tz":540},"subject":"Add doc of OAuth 2.0 Mutual-TLS Authenticate","message":"Add doc of OAuth 2.0 Mutual-TLS Authenticate\n\nThe OAuth2.0 Access Token API is modified, support to get an OAuth2.0\ncertificate-bound access token from the keystone identity server with\nOAuth 2.0 credentials and Mutual-TLS certificates.\n\nImplements: blueprint support-oauth2-mtls\nChange-Id: Ib1c391b06b167c55435360b81bb67c19a63c8b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d16cb8771d30fbceb377463dc54b74db84ab709e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d16cb8771d30fbceb377463dc54b74db84ab709e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
