)]}'
{"id":"openstack%2Fkeystone~863257","triplet_id":"openstack%2Fkeystone~stable%2Fyoga~I3ae502580588af42ac5d5f9fc6718a639b443e98","project":"openstack/keystone","branch":"stable/yoga","topic":"bug/1916662-stable/yoga","attention_set":{},"removed_from_attention_set":{"29474":{"account":{"_account_id":29474,"name":"Gheorghita BUTNARU","email":"gbutnaru@tuiasi.ro","username":"gbutnaru"},"last_update":"2024-02-07 09:42:25.000000000","reason":"Change was abandoned"},"35461":{"account":{"_account_id":35461,"name":"Alexandru Avasilcai","email":"alexandru.avasilcai@staff.tuiasi.ro","username":"ava"},"last_update":"2024-02-07 09:42:25.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3ae502580588af42ac5d5f9fc6718a639b443e98","subject":"Change error response status code in master branch","status":"ABANDONED","created":"2022-11-02 11:38:20.000000000","updated":"2024-02-07 09:42:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":845534,"cherry_pick_of_patch_set":1,"meta_rev_id":"9a7e3b0446614f1b0e301ed6862f62d088defcd8","_number":863257,"virtual_id_number":863257,"owner":{"_account_id":29474,"name":"Gheorghita BUTNARU","email":"gbutnaru@tuiasi.ro","username":"gbutnaru"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-11-02 11:53:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-02 11:38:20.000000000","updated_by":{"_account_id":29474,"name":"Gheorghita BUTNARU","email":"gbutnaru@tuiasi.ro","username":"gbutnaru"},"reviewer":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"state":"CC"},{"updated":"2022-11-02 11:53:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-07 09:42:25.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"e31d0c8cfff630afd85bf4e7e4cc906454178524","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29474,"name":"Gheorghita BUTNARU","email":"gbutnaru@tuiasi.ro","username":"gbutnaru"},"date":"2022-11-02 11:38:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad9fae2ab4e609a7291c0eaada14e2ac942ab21a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29474,"name":"Gheorghita BUTNARU","email":"gbutnaru@tuiasi.ro","username":"gbutnaru"},"date":"2022-11-02 11:39:44.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e5cda474009ef51f4c08956643e406abda34d6da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35461,"name":"Alexandru Avasilcai","email":"alexandru.avasilcai@staff.tuiasi.ro","username":"ava"},"date":"2022-11-02 11:40:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"58eaa1fcfde9a14188912c03ccf4255ca81b217e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 11:53:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92301c898122409a8f327e0125ee1f86\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8170c4078e4a49c99b55ef163298eb91 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d208b0c8b3e47869bad06fc4e2c2df4 : SUCCESS in 11m 40s","accounts_in_message":[],"_revision_number":3},{"id":"9a7e3b0446614f1b0e301ed6862f62d088defcd8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-07 09:42:25.000000000","message":"Abandoned\n\nstable/yoga branch of openstack/keystone is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/yoga if you want to further work on this patch.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"590ee620260ae72a6261c5fc9b75979eefe3e827","revisions":{"05cf8f4cde0a958cf982811509ceb814c051f2a2":{"kind":"REWORK","_number":1,"created":"2022-11-02 11:38:20.000000000","uploader":{"_account_id":29474,"name":"Gheorghita BUTNARU","email":"gbutnaru@tuiasi.ro","username":"gbutnaru"},"ref":"refs/changes/57/863257/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/57/863257/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/57/863257/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/57/863257/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/57/863257/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/57/863257/1"}}},"commit":{"parents":[{"commit":"a26b9c20de7103b1cb6cdd43bd2f2cca9a29fa2f","subject":"Merge \"Only log warnings about token length when length exceeds max_token_size\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a26b9c20de7103b1cb6cdd43bd2f2cca9a29fa2f"}]}],"author":{"name":"chenwei","email":"chen.wei@99cloud.net","date":"2022-06-13 02:59:45.000000000","tz":480},"committer":{"name":"Gheorghita BUTNARU","email":"gbutnaru@tuiasi.ro","date":"2022-11-02 11:38:18.000000000","tz":120},"subject":"Change error response status code in master branch","message":"Change error response status code in master branch\n\nClosed bug: #1916662\n\nChange-Id: I3ae502580588af42ac5d5f9fc6718a639b443e98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/05cf8f4cde0a958cf982811509ceb814c051f2a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/05cf8f4cde0a958cf982811509ceb814c051f2a2"}]},"branch":"refs/heads/stable/yoga"},"06922b460a9e8af4c284300e067a2fcc0840e99e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-11-02 11:39:44.000000000","uploader":{"_account_id":29474,"name":"Gheorghita BUTNARU","email":"gbutnaru@tuiasi.ro","username":"gbutnaru"},"ref":"refs/changes/57/863257/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/57/863257/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/57/863257/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/57/863257/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/57/863257/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/57/863257/2"}}},"commit":{"parents":[{"commit":"a26b9c20de7103b1cb6cdd43bd2f2cca9a29fa2f","subject":"Merge \"Only log warnings about token length when length exceeds max_token_size\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a26b9c20de7103b1cb6cdd43bd2f2cca9a29fa2f"}]}],"author":{"name":"chenwei","email":"chen.wei@99cloud.net","date":"2022-06-13 02:59:45.000000000","tz":480},"committer":{"name":"Gheorghita BUTNARU","email":"gbutnaru@tuiasi.ro","date":"2022-11-02 11:38:47.000000000","tz":120},"subject":"Change error response status code in master branch","message":"Change error response status code in master branch\n\nClosed bug: #1916662\n\n(cherry picked from commit cdeba521b6b28582e560c673ba1ed1ffbd25ccfc)\nChange-Id: I3ae502580588af42ac5d5f9fc6718a639b443e98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/06922b460a9e8af4c284300e067a2fcc0840e99e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/06922b460a9e8af4c284300e067a2fcc0840e99e"}]},"branch":"refs/heads/stable/yoga"},"590ee620260ae72a6261c5fc9b75979eefe3e827":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-11-02 11:40:06.000000000","uploader":{"_account_id":35461,"name":"Alexandru Avasilcai","email":"alexandru.avasilcai@staff.tuiasi.ro","username":"ava"},"ref":"refs/changes/57/863257/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/57/863257/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/57/863257/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/57/863257/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/57/863257/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/57/863257/3"}}},"commit":{"parents":[{"commit":"a26b9c20de7103b1cb6cdd43bd2f2cca9a29fa2f","subject":"Merge \"Only log warnings about token length when length exceeds max_token_size\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a26b9c20de7103b1cb6cdd43bd2f2cca9a29fa2f"}]}],"author":{"name":"chenwei","email":"chen.wei@99cloud.net","date":"2022-06-13 02:59:45.000000000","tz":480},"committer":{"name":"Alexandru Avasilcai","email":"alexandru.avasilcai@staff.tuiasi.ro","date":"2022-11-02 11:40:06.000000000","tz":0},"subject":"Change error response status code in master branch","message":"Change error response status code in master branch\n\nClosed bug: #1916662\n\nChange-Id: I3ae502580588af42ac5d5f9fc6718a639b443e98\n(cherry picked from commit cdeba521b6b28582e560c673ba1ed1ffbd25ccfc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/590ee620260ae72a6261c5fc9b75979eefe3e827"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/590ee620260ae72a6261c5fc9b75979eefe3e827"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
