)]}'
{"id":"openstack%2Fkeystone~869195","triplet_id":"openstack%2Fkeystone~master~Ib0428e9261ec5b89463b4ad845d2cc8244d95ffc","project":"openstack/keystone","branch":"master","attention_set":{"9807":{"account":{"_account_id":9807,"name":"Yaroslav","email":"yisakov@mirantis.com","username":"yarisakov"},"last_update":"2023-01-04 12:57:05.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ib0428e9261ec5b89463b4ad845d2cc8244d95ffc","subject":"Ignore non-existing domains for federation group lookup","status":"NEW","created":"2023-01-04 11:38:28.000000000","updated":"2023-01-04 12:57:05.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"65da6d35543f72c36a078b3b3f684aa1628b0d64","_number":869195,"virtual_id_number":869195,"owner":{"_account_id":9807,"name":"Yaroslav","email":"yisakov@mirantis.com","username":"yarisakov"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-01-04 12:57:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-04 11:55:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-04 12:57:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c5b1c05bcccdb78b604a5c75de80786ab90e4927","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9807,"name":"Yaroslav","email":"yisakov@mirantis.com","username":"yarisakov"},"date":"2023-01-04 11:38:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e442618d33a2baa4f37bf89910f562eac4bb18e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9807,"name":"Yaroslav","email":"yisakov@mirantis.com","username":"yarisakov"},"date":"2023-01-04 11:42:14.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"636a2d3a14ccd8ec00689de7fdb038bacf023925","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 11:55:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b1fe9952d3a4c4cb75f9a1818d9ffd5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d3e43f5341b4667afa9670513a7468a : FAILURE in 11m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a4b4f101e1ac4fcf82571d7b1958eb4b : FAILURE in 11m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"65da6d35543f72c36a078b3b3f684aa1628b0d64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 12:57:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/581288a6e2154936b8f4b072701706d5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43af622a8de645418ff03d36a94cda21 : FAILURE in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e8bdaf434c947828a5e1eea2269d87f : FAILURE in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d2e9c0893914dc9b733c3ef13aabdc0 : FAILURE in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3385c01adaff43c68cc354f1c726800d : FAILURE in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c1b133e78b144dd9aa6d993c1f981957 : FAILURE in 3m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/581aa59f31eb4776b57389a6875e258b : FAILURE in 5m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/9e165d56d0fa40cdb975531a70406add : SUCCESS in 52m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e5ff8eedfac74c8c94c2c9204eb997d4 : FAILURE in 1h 13m 24s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/03efb68493934a0ca66fd3a4549e835e : SUCCESS in 25m 23s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/2dfa1d5490c541139c729c566f5ecf64 : SUCCESS in 27m 53s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6089c15adcc4672a5a20e1e1f696e95 : SUCCESS in 25m 28s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/25cfc547bb1f42daa880bf6cfba91735 : SUCCESS in 23m 08s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/27bf71f812af4912933dbdc3717537a0 : SUCCESS in 9m 06s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/e33cd6e8371a4f0cb103646f395c7aeb : FAILURE in 23m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee89e26d80974ac1af325935d2eb6a0a : SUCCESS in 40m 30s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/e5850a8100e64ec693d3ffdc35392c17 : SUCCESS in 54m 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"eac0dcaf379d8e5b24c8ec9ddaf5f2b5b4c012d1","revisions":{"5f5c79f4926c010b50ae69f02777265a85f97ec5":{"kind":"REWORK","_number":1,"created":"2023-01-04 11:38:28.000000000","uploader":{"_account_id":9807,"name":"Yaroslav","email":"yisakov@mirantis.com","username":"yarisakov"},"ref":"refs/changes/95/869195/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/95/869195/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/95/869195/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/95/869195/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/95/869195/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/95/869195/1"}}},"commit":{"parents":[{"commit":"7d4047cb69139c22f4e321e9ba8b287a47168df2","subject":"Merge \"requirements: Bump linter requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7d4047cb69139c22f4e321e9ba8b287a47168df2"}]}],"author":{"name":"Yaroslav Isakov","email":"yaroslav.isakov@gmail.com","date":"2023-01-04 11:38:16.000000000","tz":60},"committer":{"name":"Yaroslav Isakov","email":"yaroslav.isakov@gmail.com","date":"2023-01-04 11:38:16.000000000","tz":60},"subject":"Ignore non-existing domains for federation group lookup","message":"Ignore non-existing domains for federation group lookup\n\nCurrent code is ignoring group from federation mapping, if they doesn\u0027t exist,\nbut errors with keystone.exception.DomainNotFound: Could not find domain error,\nif domain from mapping doesn\u0027t exist. Let\u0027s ignore this one, too.\n\nChange-Id: Ib0428e9261ec5b89463b4ad845d2cc8244d95ffc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5f5c79f4926c010b50ae69f02777265a85f97ec5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5f5c79f4926c010b50ae69f02777265a85f97ec5"}]},"branch":"refs/heads/master"},"eac0dcaf379d8e5b24c8ec9ddaf5f2b5b4c012d1":{"kind":"NO_CHANGE","_number":2,"created":"2023-01-04 11:42:14.000000000","uploader":{"_account_id":9807,"name":"Yaroslav","email":"yisakov@mirantis.com","username":"yarisakov"},"ref":"refs/changes/95/869195/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/95/869195/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/95/869195/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/95/869195/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/95/869195/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/95/869195/2"}}},"commit":{"parents":[{"commit":"7d4047cb69139c22f4e321e9ba8b287a47168df2","subject":"Merge \"requirements: Bump linter requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7d4047cb69139c22f4e321e9ba8b287a47168df2"}]}],"author":{"name":"Yaroslav Isakov","email":"yisakov@mirantis.com","date":"2023-01-04 11:42:01.000000000","tz":60},"committer":{"name":"Yaroslav Isakov","email":"yisakov@mirantis.com","date":"2023-01-04 11:42:01.000000000","tz":60},"subject":"Ignore non-existing domains for federation group lookup","message":"Ignore non-existing domains for federation group lookup\n\nCurrent code is ignoring group from federation mapping, if they doesn\u0027t exist,\nbut errors with keystone.exception.DomainNotFound: Could not find domain error,\nif domain from mapping doesn\u0027t exist. Let\u0027s ignore this one, too.\n\nChange-Id: Ib0428e9261ec5b89463b4ad845d2cc8244d95ffc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/eac0dcaf379d8e5b24c8ec9ddaf5f2b5b4c012d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/eac0dcaf379d8e5b24c8ec9ddaf5f2b5b4c012d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
