)]}'
{"id":"openstack%2Fkeystone~870619","triplet_id":"openstack%2Fkeystone~master~Idfd47a788d481f57a9681c1b6f65bc7d6d7bda06","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2023-01-20 15:35:28.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idfd47a788d481f57a9681c1b6f65bc7d6d7bda06","subject":"DNM - Test CI","status":"ABANDONED","created":"2023-01-16 15:58:01.000000000","updated":"2023-01-20 15:35:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"394dc3d468fada782f87c9738605be736f4e4a32","_number":870619,"virtual_id_number":870619,"owner":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-01-16 17:25:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-16 17:25:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a712785545f230a225a64e795807745abfb4582e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2023-01-16 15:58:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5daafaaf0f0f1be6e45db28a9c946329fd20750e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 17:25:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/56bd1d71c1aa42139133781cb28715d8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5e64b46cf3f49a28ee96978ac703986 : FAILURE in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da71b8fbd2fe41c3bfeec61fcdb773fb : FAILURE in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb6f6a02c4d54ac5823fbbd754593c90 : FAILURE in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4942232ba9224ffcae3f5289f6baaba0 : FAILURE in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/37deffcddad0452dbc1f66b8276ec903 : FAILURE in 2m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaa169834324429f9ab2c374cc3ecef4 : FAILURE in 3m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/326b1beb52214685a10588b61d7f3830 : SUCCESS in 54m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8e8c0f4dacef468eb071ee8cdc9d30ec : SUCCESS in 1h 24m 35s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/c1a31e0333574ae3b687485c51d76d1e : SUCCESS in 18m 51s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/f5c19f7e22a641a08074ea5dbe13ec51 : SUCCESS in 45m 31s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df56b872880d483d8341537bd121d3c1 : SUCCESS in 36m 17s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/df95c5af69ad4de6948712337dbc254d : SUCCESS in 31m 13s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/8539e79b7dd349ca94023faab74cd59b : SUCCESS in 9m 25s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/34f47ed52f054505ba1048657828ecd5 : FAILURE in 21m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07af5321e1804038984e90b5f2a15e95 : SUCCESS in 56m 01s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/eeac6f2956f54886a7b0b54743f8fab4 : SUCCESS in 28m 13s","accounts_in_message":[],"_revision_number":1},{"id":"b993bb1f7a5f3aacfa03168fc2a1b3be3ea3395e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 18:31:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f7e1e7d513f41c2bd5ed57d123b386a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dd65e6c833724a3089350065ab8485c8 : FAILURE in 8m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b462ac0ed17e4de3903f62de90c348eb : FAILURE in 9m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"394dc3d468fada782f87c9738605be736f4e4a32","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2023-01-20 15:35:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"59d036a5d912ba56e27215ce1244e14c6a1d2c14","revisions":{"59d036a5d912ba56e27215ce1244e14c6a1d2c14":{"kind":"REWORK","_number":1,"created":"2023-01-16 15:58:01.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/19/870619/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/19/870619/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/19/870619/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/19/870619/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/19/870619/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/19/870619/1"}}},"commit":{"parents":[{"commit":"7d4047cb69139c22f4e321e9ba8b287a47168df2","subject":"Merge \"requirements: Bump linter requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7d4047cb69139c22f4e321e9ba8b287a47168df2"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2023-01-16 15:57:32.000000000","tz":0},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2023-01-16 15:57:32.000000000","tz":0},"subject":"DNM - Test CI","message":"DNM - Test CI\n\nChange-Id: Idfd47a788d481f57a9681c1b6f65bc7d6d7bda06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/59d036a5d912ba56e27215ce1244e14c6a1d2c14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/59d036a5d912ba56e27215ce1244e14c6a1d2c14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
