)]}'
{"id":"openstack%2Fkeystone~874624","triplet_id":"openstack%2Fkeystone~stable%2Fussuri~I7fd62655a46b2c029f4c386a1c1325a0099ed0a7","project":"openstack/keystone","branch":"stable/ussuri","hashtags":[],"change_id":"I7fd62655a46b2c029f4c386a1c1325a0099ed0a7","subject":"DNM: testing tempest-full-py3 fix","status":"ABANDONED","created":"2023-02-21 21:35:08.000000000","updated":"2023-02-22 15:46:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":874623,"cherry_pick_of_patch_set":1,"meta_rev_id":"5366614467178c333317d8599cb92cc7ed9f1a64","_number":874624,"virtual_id_number":874624,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-02-22 00:22:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-22 00:22:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"75fc4fa9a2afd3fef347a8645541c24db9bda458","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-21 21:35:08.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"8cf32a9c439cc59b6e353669055040da03e29b69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 00:22:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70a62c0837a042748091755247a36b11\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a66ab5b59b94fc28ea5107a5d3e50c5 : SUCCESS in 14m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18ffed0eb28c410185ed2ac21a68e654 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fffddc05e44242cbaaca7fe5ec71f104 : SUCCESS in 12m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/91f1e45cab5a4bd182d1f9d98836c981 : SUCCESS in 15m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/310804751d4742afbce1ec84c241c29a : SUCCESS in 16m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/10cceee9c1e3469786bd185b2c3fb339 : SUCCESS in 2h 39m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ffd07e9efd0644098a3b09154a9b29d3 : SUCCESS in 1h 41m 59s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/8fee74fcbba9489f80be19417b2f705a : SUCCESS in 1h 10m 38s\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/9e354fb1e2ee40daad6142e16fe20f14 : FAILURE in 40m 33s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/14200732dc9f4cfdbfa59fd4f8775452 : FAILURE in 48m 17s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/0c4de8a8ecdd4e088fd4f7fb995de793 : FAILURE in 18m 18s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ca9c51b341c74a5d9e0b73dea08a5509 : SUCCESS in 49m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be9845d0ff3d486dba8ac4ed823a1669 : SUCCESS in 58m 40s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/dde882bdd6c040caaf91fe4e20c0b5aa : SUCCESS in 39m 22s","accounts_in_message":[],"_revision_number":1},{"id":"5366614467178c333317d8599cb92cc7ed9f1a64","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-22 15:46:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"26ce30ad1b0fb31d802a47701c7cf75d81d3be20","revisions":{"26ce30ad1b0fb31d802a47701c7cf75d81d3be20":{"kind":"REWORK","_number":1,"created":"2023-02-21 21:35:08.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/24/874624/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/24/874624/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/874624/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/874624/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/24/874624/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/24/874624/1"}}},"commit":{"parents":[{"commit":"1ab860a08e527ca9e0c82a49fbf004d415fec991","subject":"Fix host:port handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1ab860a08e527ca9e0c82a49fbf004d415fec991"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-02-21 21:32:54.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-02-21 21:35:08.000000000","tz":0},"subject":"DNM: testing tempest-full-py3 fix","message":"DNM: testing tempest-full-py3 fix\n\nDepends-ON: https://review.opendev.org/c/openstack/tempest/+/874704\nChange-Id: I7fd62655a46b2c029f4c386a1c1325a0099ed0a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/26ce30ad1b0fb31d802a47701c7cf75d81d3be20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/26ce30ad1b0fb31d802a47701c7cf75d81d3be20"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
