)]}'
{"id":"openstack%2Fkeystone~875540","triplet_id":"openstack%2Fkeystone~master~Id4c5e80d665fa90d07517bc58c26b44d543d34b3","project":"openstack/keystone","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-06-22 23:23:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id4c5e80d665fa90d07517bc58c26b44d543d34b3","subject":"tests: Enable SQLAlchemy 2.0 deprecation warnings","status":"MERGED","created":"2023-02-27 17:44:26.000000000","updated":"2023-06-22 23:25:15.000000000","submitted":"2023-06-22 23:23:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875540-sqlalchemy-20","meta_rev_id":"686c6c494975f9e96ea5dbcd378048c5a7027625","_number":875540,"virtual_id_number":875540,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-22 23:23:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2023-06-22 21:28:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2023-05-29 21:00:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-06-19 08:23:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2023-06-22 21:28:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-27 20:22:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-29 21:00:26.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-06-19 08:23:34.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2023-06-22 21:28:44.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"}],"messages":[{"id":"4d2dafb43f20f0f4c49f9d8f1d3acceb33894b02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-27 17:44:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef76de660ac71f6b5395164e6ef6d904125c2c9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 20:22:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/700c093c8c874663bb6100f64a7ddf48\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5e67cc3590b34ee99350c0f3e9dd31de : SUCCESS in 33m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7dd31dd380c438b824e814c1e575196 : SUCCESS in 40m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5df2caf8a9124e3db180198fb9fc5320 : SUCCESS in 24m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc1e13654acb4c35b442df55c42c0447 : SUCCESS in 31m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/498cc0509fb045bea86790421ba1626f : SUCCESS in 31m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/036cc9a28cf641dba4459901e3086c26 : SUCCESS in 20m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/1a7244d9aa3344f0ba1be662972233cd : SUCCESS in 1h 07m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1765ad9a20ae40cc942747b6ae6e7c50 : SUCCESS in 2h 04m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fae3ebcf42fd4027a7adb81a9ece84cf : SUCCESS in 37m 38s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/9078237151394986933e0ce0a594bbd4 : SUCCESS in 57m 53s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/b391b2f3c1f444fcae321b4a3cbf01db : SUCCESS in 1h 20m 33s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0f9c5bd547d41f294b260b153acb5ff : SUCCESS in 25m 41s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/4a5ece94a1eb496eab2db516c9ddb4af : SUCCESS in 42m 17s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/4ab28c24669c468187ac310db9522295 : SUCCESS in 24m 34s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/f3532ee60daf4ca8b1da7a34c38d4484 : FAILURE in 37m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbd9e6960e1b496b8ba0a008588b1dca : SUCCESS in 56m 01s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/b54a893b046a41b589ae08f37c10a7fb : SUCCESS in 42m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a53822780dbe2d2aad7027f64d3ab4a23df1b2ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-28 17:26:46.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"05c1ce0cf70b12ca59ee64f1b6ab1de27725a028","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 17:41:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f29fcb180dfb40cfb681b18ff67b4351\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d9f138c56444438b7cf8dcf4643b23c : SUCCESS in 13m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a1ba5e013454612ac943dc6dfc603b5 : SUCCESS in 12m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8d1a78038f1b350cbbd2d7f6be0ee2ae964c4e7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 20:57:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f146a217dfc8455b829c5a3b467afecc\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3491c2240129428e841ff52c9b5e7044 : SUCCESS in 4m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3458557f9ede4553adc8ba33ef4da74d : SUCCESS in 13m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4037c54f68d04793a2fa93d7c784770e : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b532b32db0245e69a79df05412501f7 : SUCCESS in 11m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f14b7e003f994785992728160ee52420 : SUCCESS in 10m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f18874e0c2094db288d1e0bb4d77e597 : SUCCESS in 11m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/02acfe7f39334d2d9f322ef5f3565c9b : SUCCESS in 1h 05m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d3a80e962745425d951dc08a2c3c7318 : SUCCESS in 1h 25m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/040e8ebcdb724a2e909e1e401745990c : SUCCESS in 6m 38s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/2189daa9ae6d48ba95882939e5fedf97 : SUCCESS in 43m 16s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/4766aaa2a5de41e7a26ada5c9b51001d : SUCCESS in 49m 56s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/521e98e2ead04642a5a0dca89968546e : SUCCESS in 32m 30s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/50d233484d3d422da7ba97f7714ee256 : SUCCESS in 28m 10s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/572f77a6b98c47c59e627866c57bab2d : SUCCESS in 9m 50s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/2b8aa115832343099c72c5687ac23de8 : FAILURE in 26m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4a5d8061b1e4d98bf14d17267489438 : SUCCESS in 45m 17s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/dccacde09ead4772b8631c6d5f6e941c : SUCCESS in 35m 20s","accounts_in_message":[],"_revision_number":2},{"id":"6730011d8eb5bf3d2d93b240db78c4389977f4ee","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-05-29 21:00:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"80da84f37e9e77d41066b9542dd5c82579566617","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-06-19 08:23:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"35bafce16c256b196bb138691e6c546e74d40dd0","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2023-06-22 21:28:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"05da8195b79867cea37e728b8f0a914fc9e31c03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 21:29:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"61211b1f108d429b7dc51d99e391847368c5cd0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 23:23:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2eb45c6e8d79410db82a26603a4e8b59\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/31ee3204ce444562b9b1353246c4e81e : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/013278fb963c4f429f48468cb3cc2d9d : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3849bea015b4edcb2f52367d506411e : SUCCESS in 16m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff1425f1c5da46efaf7b03b02e15764f : SUCCESS in 11m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efa20a5ddf5c4723a0cc19e3c22ab5b8 : SUCCESS in 12m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dd3b0538be74fa18100bb0e0803ccaa : SUCCESS in 12m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/003061322f644a59b51f020f4692dc9e : SUCCESS in 1h 05m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ffb906bc1b0d49d5a498fd128cbe0c4c : SUCCESS in 1h 50m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9cd27c67f4804a68b683bd451ddc1f4c : SUCCESS in 7m 07s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/4ec9d62f5ada4f8cab587a022d9ff8b9 : SUCCESS in 23m 03s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/baa5d191566f477ca2d8dcc077664a09 : SUCCESS in 27m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed07c830edc8490aa030e4b14ab6b1c0 : SUCCESS in 54m 36s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/25d6db0bfa38414fb1042be8ead5f4ed : SUCCESS in 31m 42s","accounts_in_message":[],"_revision_number":2},{"id":"a9474dd0e2df2c472d1ae995e1c78400e7e9fcce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 23:23:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"686c6c494975f9e96ea5dbcd378048c5a7027625","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 23:25:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/962d98e192d4411cb6c263a39bfbd489\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/d4dd39312fd9426b87fe1bcf14956f10 : SUCCESS in 38s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8ffb1c5e75d48118e4b2cf99b2ddc0a : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/499efc8dee034f6bb5fd4ee2c1e703ef : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"802539ad41718b12cfd4dd636121d1f3b1d07ffc","revisions":{"a1e270eca757005b0bd9ff6ee4e172357a5b2a57":{"kind":"REWORK","_number":1,"created":"2023-02-27 17:44:26.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/875540/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/40/875540/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/40/875540/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/40/875540/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/40/875540/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/40/875540/1"}}},"commit":{"parents":[{"commit":"6b7de4e0e506e5218cceebbdbfbf66ea12c31056","subject":"Bump SQLAlchemy minimum version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6b7de4e0e506e5218cceebbdbfbf66ea12c31056"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-04-20 16:49:41.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-02-27 15:55:40.000000000","tz":0},"subject":"tests: Enable SQLAlchemy 2.0 deprecation warnings","message":"tests: Enable SQLAlchemy 2.0 deprecation warnings\n\nWell, sort of. We enable them but immediately filter out the ones we\u0027re\nactually seeing, the rationale being that we can address these in a\npiecemeal fashion without the risk of introducing new issues.\n\nThere\u0027s a lot more to be done here. However, the work done in oslo.db\nand other projects [1] should provide a guide for how to resolve the\noutstanding issues.\n\n[1] https://review.opendev.org/q/topic:%2522sqlalchemy-20%2522\n\nChange-Id: Id4c5e80d665fa90d07517bc58c26b44d543d34b3\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a1e270eca757005b0bd9ff6ee4e172357a5b2a57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a1e270eca757005b0bd9ff6ee4e172357a5b2a57"}]},"branch":"refs/heads/master"},"802539ad41718b12cfd4dd636121d1f3b1d07ffc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-28 17:26:46.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/875540/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/40/875540/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/40/875540/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/40/875540/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/40/875540/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/40/875540/2"}}},"commit":{"parents":[{"commit":"f24c77c774bb4c76f1ed85d814766eb559ba3571","subject":"Bump SQLAlchemy minimum version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f24c77c774bb4c76f1ed85d814766eb559ba3571"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-04-20 16:49:41.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-02-28 17:26:06.000000000","tz":0},"subject":"tests: Enable SQLAlchemy 2.0 deprecation warnings","message":"tests: Enable SQLAlchemy 2.0 deprecation warnings\n\nWell, sort of. We enable them but immediately filter out the ones we\u0027re\nactually seeing, the rationale being that we can address these in a\npiecemeal fashion without the risk of introducing new issues.\n\nThere\u0027s a lot more to be done here. However, the work done in oslo.db\nand other projects [1] should provide a guide for how to resolve the\noutstanding issues.\n\n[1] https://review.opendev.org/q/topic:%2522sqlalchemy-20%2522\n\nChange-Id: Id4c5e80d665fa90d07517bc58c26b44d543d34b3\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/802539ad41718b12cfd4dd636121d1f3b1d07ffc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/802539ad41718b12cfd4dd636121d1f3b1d07ffc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
