)]}'
{"id":"openstack%2Fkeystone~875761","triplet_id":"openstack%2Fkeystone~master~I98e2b17debc509a8db5a424eb5d602c7c075d558","project":"openstack/keystone","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-06-27 15:41:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I98e2b17debc509a8db5a424eb5d602c7c075d558","subject":"db: Don\u0027t pass strings to \u0027Connection.execute\u0027","status":"MERGED","created":"2023-02-28 17:26:46.000000000","updated":"2023-06-27 15:43:07.000000000","submitted":"2023-06-27 15:41:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875761-sqlalchemy-20","meta_rev_id":"291910120fa7c14d0e2eed1bf5da91d37bd64793","_number":875761,"virtual_id_number":875761,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-27 15:41:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2023-06-22 21:33:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2023-04-04 23:48:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2023-06-27 13:11:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2023-05-29 21:03:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-06-26 07:50:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2023-06-27 13:11:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-28 18:32:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-28 20:39:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-04 23:48:22.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2023-05-29 21:03:04.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-06-22 21:33:41.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2023-06-26 07:50:40.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-06-27 13:11:23.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"347c2411fcad9f3be28981e1b4a0e37d16493e29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-28 17:26:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ee61ccefdce3af97e8adf32fa3fbc12f2aa4cce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 18:32:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b9b3ae8bfb7483cb40761c7264f58e9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d92a3a2ba71d49ff8cbd7a272b2ef119 : TIMED_OUT in 1h 01m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e25dee89433d45a6bb3629cac14533c8 : TIMED_OUT in 1h 02m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e650065e9ed54bc60af5b6d88c0fcb53d5d725a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 20:39:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8001c2e937647aeb21a612a5b8eb831\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/237ec3d2d2464a90896163e02a1de442 : SUCCESS in 16m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c600574a424c45879100fa00ad165c6d : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a46ab26f5b684585be659811f2ede532 : SUCCESS in 11m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9078a496ba44949be22aba475a091a6 : SUCCESS in 10m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/235021d700c94836849e4abb4f312de5 : SUCCESS in 13m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/0079d4b1f72a4fd9893a7e141c98cd84 : SUCCESS in 56m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ca862297a67d4e6aa40bbfbee5c8e074 : SUCCESS in 1h 15m 49s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/118919386a874da89831a3afa3f2384b : SUCCESS in 36m 55s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/9f1f610c961b49ff8ea63a70bc56d815 : SUCCESS in 41m 12s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1a6d6d9c0a246a58a3b3ea59274e0b9 : SUCCESS in 31m 50s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/543422539406436a870b9ebb7963d130 : SUCCESS in 22m 42s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/53d0baf072b645928da81c391042fb2e : SUCCESS in 9m 31s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ff634f422eff425f94ec424cda782f0d : FAILURE in 12m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8794a0fc7b943188487afa3c68ed398 : SUCCESS in 50m 19s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/1cd40e169eaf4a7aa67a87f1451c7c19 : SUCCESS in 32m 14s","accounts_in_message":[],"_revision_number":1},{"id":"3e8ead6dee61e835b4da904bc4fb183ee271e2e8","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-04-04 23:48:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8bc368d74c8a289cd64ebbdf1d032c443dc36eaa","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-05-29 21:03:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7bb12cf572638a6ee3789f571c5849f65e6753ab","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2023-06-22 21:33:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"01b30fd9c5b5304a6c4f704c5d3b11eeb1538ba1","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-06-26 07:50:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7866b62741c386781ca686e5d6ad0a0cc40851a1","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-06-27 13:11:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f338c9d896514b764cf1b877a8094ad00b4e1e7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 13:11:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0d611dedcc88547bc0fb57b8441c607d183a13f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 15:41:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f5efa6e84564710a3d3a4275db28b77\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e7e2a49f1a64628a76fa4a93d2e2c77 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2473de896dec4d85a83c88776728f295 : SUCCESS in 13m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d270e864e37d44b79d1b90d49b9973bf : SUCCESS in 14m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ae180d0d4664dca8a13f96c5d268b4a : SUCCESS in 19m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90d99b09fda54e88b0cbb5f27202e058 : SUCCESS in 10m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/2e88050412254c95b2d2b27048c98637 : SUCCESS in 1h 21m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc5e2d5081664bb7bcb52e18aa73c846 : SUCCESS in 1h 26m 27s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/6c1c9e51bce34423ae59c84688f4543d : SUCCESS in 22m 46s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/bf4c5d44db7845f6b3d9a6306606d967 : SUCCESS in 21m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5434e59003c4813ad2c31f58eadb4de : SUCCESS in 42m 11s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/cc0e405ce0bc4e54a20223ed80cc861e : SUCCESS in 35m 29s","accounts_in_message":[],"_revision_number":1},{"id":"2e51fe983b9f865c4b0eaf260ac4d49824fc23db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 15:41:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"291910120fa7c14d0e2eed1bf5da91d37bd64793","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 15:43:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2262e7073f824e6d8fa91f057fd0c396\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0732dbe4e7648768c15e5e15c0eb3bc : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9165c67c10a4aac562de84cb45b4f2872128d98a","revisions":{"9165c67c10a4aac562de84cb45b4f2872128d98a":{"kind":"REWORK","_number":1,"created":"2023-02-28 17:26:46.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/61/875761/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/61/875761/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/61/875761/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/61/875761/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/61/875761/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/61/875761/1"}}},"commit":{"parents":[{"commit":"d9b447bacd45e8adec4b8ca226273c0967cd63dc","subject":"db: Replace use of \u0027autoload\u0027 parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d9b447bacd45e8adec4b8ca226273c0967cd63dc"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-02-28 12:05:38.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-02-28 17:26:39.000000000","tz":0},"subject":"db: Don\u0027t pass strings to \u0027Connection.execute\u0027","message":"db: Don\u0027t pass strings to \u0027Connection.execute\u0027\n\nResolve the following RemovedIn20Warning warning:\n\n  Passing a string to Connection.execute() is deprecated and will be\n  removed in version 2.0.  Use the text() construct, or the\n  Connection.exec_driver_sql() method to invoke a driver-level SQL\n  string.\n\nChange-Id: I98e2b17debc509a8db5a424eb5d602c7c075d558\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9165c67c10a4aac562de84cb45b4f2872128d98a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9165c67c10a4aac562de84cb45b4f2872128d98a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
