)]}'
{"id":"openstack%2Fkeystone~875765","triplet_id":"openstack%2Fkeystone~master~Ib3f9099160265c4eafea1b2e38537c58eadf9a5c","project":"openstack/keystone","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-06-27 16:02:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib3f9099160265c4eafea1b2e38537c58eadf9a5c","subject":"tests: Rework BannedDBSchemaOperations fixture","status":"MERGED","created":"2023-02-28 17:26:46.000000000","updated":"2023-06-27 16:04:29.000000000","submitted":"2023-06-27 16:02:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875765-sqlalchemy-20","meta_rev_id":"8cdaa0c4050b89d45ac317209d94df6a5d341b08","_number":875765,"virtual_id_number":875765,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-27 16:02:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2023-06-22 21:49:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2023-06-23 14:04:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2023-06-23 14:04:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-28 17:42:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-28 20:18:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-22 21:49:44.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2023-06-23 14:04:28.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"479f5d5b87b26a44652faae074612219019aebe1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-28 17:26:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c28d78a241993da60c09952eadc514bd9130223","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 17:42:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b612ce68d94f496b8ded4082d8d271c9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e6e8af1d1aed46a8bd3b21026ced575a : SUCCESS in 11m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f9a2a40be341499cbe553bbee05b6ade : SUCCESS in 12m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"95c4762530711c76864c29ad8186225a96c1d07a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-28 17:55:18.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"206a1dd6540516e8d8acf83447c05c2dacdb93de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 18:59:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb5c423a47b749679d09a068e53909f9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0549aaf6742c46d09ed50a766858b901 : SUCCESS in 14m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/af65fc40454b4d1a9e355339b8c44cad : TIMED_OUT in 1h 01m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"15f7794cffe0ee3dfa46946b46d587a4e88c579b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 20:18:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2240ced5361b4488845c555ae3e290f4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c84b605b89d84053bad23fbeb75dcc1d : SUCCESS in 13m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e252043714044dbb240d8bfcfad7ce1 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09bd2cecc3024e80a133274fd660cc6d : SUCCESS in 13m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/216ee32fc857436bb9dcc113332ad3b9 : SUCCESS in 12m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/083e0877643940acbb6324a9ebf53f1d : SUCCESS in 12m 22s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/bcdea302eb4345979a491b39ed98e208 : SUCCESS in 33m 28s","accounts_in_message":[],"_revision_number":2},{"id":"4d126ac24400217f7a49747a47b919fa10991c77","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2023-06-22 21:49:44.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9c234c60de679335cf7728573df632aac1bdf6b5","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-06-23 14:04:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"67fc357a188b3e373b012059ae4723c5f6230351","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 13:12:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cd43a1921e9d6b6cfe19a2be2f33762a8fd6ad47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 16:02:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9e9c6b6e26b411fb08255797d7c043e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9a50ddb89e24b7ea19c6bd2a097cb79 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a958f94a891e41ad95c9e3b99ece7a81 : SUCCESS in 10m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39af6c0658f1473a995aed88cb2fa5e3 : SUCCESS in 12m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/536e4d70d14d495db3324570314124b7 : SUCCESS in 9m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fd17498c93749cc913d32b4b511b7fb : SUCCESS in 11m 04s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/3d4dee2b26b54d8a8ac2584fbc56fa78 : SUCCESS in 34m 26s","accounts_in_message":[],"_revision_number":2},{"id":"54f073a921e8d3b232fe4934dbb91d053db1e369","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 16:02:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8cdaa0c4050b89d45ac317209d94df6a5d341b08","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 16:04:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a75def7734e4ea48f302b183e3d64a8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/224c6fbb415b4ce1a8d92d6ed21fbaa5 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5e9f32469e1ebf467e2713d733490b9ba623cd04","revisions":{"e0527b2f5c4b1703957fbc663a589c767b542919":{"kind":"REWORK","_number":1,"created":"2023-02-28 17:26:46.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/875765/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/65/875765/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/65/875765/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/65/875765/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/65/875765/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/65/875765/1"}}},"commit":{"parents":[{"commit":"8c9462f6fa65fb615af4d4bfe8afae0d0143a644","subject":"db: Remove legacy migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8c9462f6fa65fb615af4d4bfe8afae0d0143a644"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-02-28 17:10:47.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-02-28 17:26:39.000000000","tz":0},"subject":"tests: Rework BannedDBSchemaOperations fixture","message":"tests: Rework BannedDBSchemaOperations fixture\n\nTwo issues here. The most pressing is that we are banning use of the\n\u0027batch_alter_table\u0027 operation. There\u0027s no reason to do this as it isn\u0027t\nan operation per se, but rather a way of modifying how operations run.\n\nFixing this highlighted a weird error case, whereby the error message\nindicated that we were calling a different operation from the one we\nwere *actually* calling. This made fixing the issue far more time\nconsuming than it should have been. I think this is because of how\nPython is interpreting lambdas but I\u0027m not certain. The function is\nrewritten to avoid the issue, simplifying it significantly in the\nprocess.\n\nChange-Id: Ib3f9099160265c4eafea1b2e38537c58eadf9a5c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e0527b2f5c4b1703957fbc663a589c767b542919"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e0527b2f5c4b1703957fbc663a589c767b542919"}]},"branch":"refs/heads/master"},"5e9f32469e1ebf467e2713d733490b9ba623cd04":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-02-28 17:55:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/875765/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/65/875765/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/65/875765/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/65/875765/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/65/875765/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/65/875765/2"}}},"commit":{"parents":[{"commit":"8c9462f6fa65fb615af4d4bfe8afae0d0143a644","subject":"db: Remove legacy migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8c9462f6fa65fb615af4d4bfe8afae0d0143a644"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-02-28 17:10:47.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-02-28 17:46:53.000000000","tz":0},"subject":"tests: Rework BannedDBSchemaOperations fixture","message":"tests: Rework BannedDBSchemaOperations fixture\n\nTwo issues here. The most pressing is that we are banning use of the\n\u0027batch_alter_table\u0027 operation. There\u0027s no reason to do this as it isn\u0027t\nan operation per se, but rather a way of modifying how operations run.\n\nFixing this highlighted a weird error case, whereby the error message\nwe\u0027d see when we called a banned operation would indicate that we had\ncalled a different operation from the one we were *actually* calling.\nThis made fixing the issue far more time consuming than it should have\nbeen (I thought it was doing something else entirely!). Turns out this\nis due to late binding [1]. The fixture is rewritten to avoid the issue,\nsimplifying it significantly in the process.\n\n[1] https://stackoverflow.com/a/3431699/613428\n\nChange-Id: Ib3f9099160265c4eafea1b2e38537c58eadf9a5c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5e9f32469e1ebf467e2713d733490b9ba623cd04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5e9f32469e1ebf467e2713d733490b9ba623cd04"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
