)]}'
{"id":"openstack%2Fkeystone~878187","triplet_id":"openstack%2Fkeystone~master~I903db2fd2ac810ec96dbd25fc6529752c08f9a79","project":"openstack/keystone","branch":"master","topic":"bug1998789-fixup","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2023-05-16 15:11:37.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}}},"hashtags":[],"change_id":"I903db2fd2ac810ec96dbd25fc6529752c08f9a79","subject":"[PooledLDAPHandler] Clean up the fix for result3()","status":"MERGED","created":"2023-03-22 03:59:16.000000000","updated":"2023-05-16 18:08:58.000000000","submitted":"2023-05-16 18:07:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878187-bug1998789-fixup","meta_rev_id":"384951a432f08f0cf033d4e241c705f3ae2ba1c9","_number":878187,"virtual_id_number":878187,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":34980,"name":"Mustafa Kemal Gilor","email":"mustafa.gilor@canonical.com","username":"mustafakemalgilor"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-16 18:07:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2023-05-05 14:06:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2023-04-17 12:05:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34980,"name":"Mustafa Kemal Gilor","email":"mustafa.gilor@canonical.com","username":"mustafakemalgilor"},{"value":1,"date":"2023-03-22 14:23:02.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2023-03-29 12:04:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-03-24 15:09:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-26 00:47:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2023-05-05 14:06:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":34980,"name":"Mustafa Kemal Gilor","email":"mustafa.gilor@canonical.com","username":"mustafakemalgilor"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34980,"name":"Mustafa Kemal Gilor","email":"mustafa.gilor@canonical.com","username":"mustafakemalgilor"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-22 04:09:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-22 05:30:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-22 14:23:02.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2023-03-24 15:09:54.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2023-03-26 00:47:50.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2023-03-29 12:04:06.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-04-17 12:05:46.000000000","updated_by":{"_account_id":34980,"name":"Mustafa Kemal Gilor","email":"mustafa.gilor@canonical.com","username":"mustafakemalgilor"},"reviewer":{"_account_id":34980,"name":"Mustafa Kemal Gilor","email":"mustafa.gilor@canonical.com","username":"mustafakemalgilor"},"state":"REVIEWER"},{"updated":"2023-05-05 14:06:53.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"}],"messages":[{"id":"2ffc80b1befb97fd0798c226e323e1998d7493cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-03-22 03:59:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae262967ac8b321bd52e613a982a7f4cf9ab5d5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 04:09:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2500d3a078604dcc99cb0a6bd98393ec\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/36f64f50541b4a96a2cfd83c7142571f : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0f04d2dc190845c9a8bc9144ba68c1cd : SUCCESS in 9m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dc981d19f21fc110dc71d778a7fdc754172eafd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 05:30:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8910f6ca556e40809015527da019595f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e931dd68224b433bbcbbb7f827a5eb98 : SUCCESS in 15m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca80a091008b4206b185f4ee50a1d9e2 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/599c06ca64c74a23a21339ae69846f17 : SUCCESS in 10m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d1ac7191a6c0461cb9fd7ce85594482e : SUCCESS in 12m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/436938853982404bb89fa28a526f6fca : SUCCESS in 12m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/8b3b9e4055de49d9b8ae241e5f7db362 : SUCCESS in 45m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/921f5c95a1a44e7b8a4a0192ef248eb1 : SUCCESS in 1h 25m 26s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/9dd31314bcc84317abd9905e3d95d707 : SUCCESS in 34m 41s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/cbe1c4cc8bac46479ca1338d4c1fe97e : SUCCESS in 44m 00s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b94b19cbbd004ae2b51a27d53aa1b280 : SUCCESS in 41m 05s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/4245815b9262422088af733fd6196662 : SUCCESS in 40m 00s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/ad79c8a1e09a44779b30084ee35ba8cf : SUCCESS in 23m 25s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/377de828c304476f9146fe5996a04632 : FAILURE in 21m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89c99962e8644ae0b58eaf3715656b20 : SUCCESS in 59m 58s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/83964d49d55345e7817327d6f7e5dbda : SUCCESS in 32m 04s","accounts_in_message":[],"_revision_number":1},{"id":"5f818503d04252c1fb6b165700b3b9ba7833b835","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-03-22 14:23:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"64c42da6539fe2c98e727e40349478646c660515","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-03-24 15:09:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"73dad59ed6cceb66b1bac5c3ce61ace8e78242b9","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-03-26 00:47:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1e46d3c89d717a8126b102743b2cf829e84c6621","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-03-29 12:04:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"397f18139510d98c83feabba9bee55c3627ccd32","author":{"_account_id":34980,"name":"Mustafa Kemal Gilor","email":"mustafa.gilor@canonical.com","username":"mustafakemalgilor"},"date":"2023-04-17 12:05:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9d1db1fcf8ab7da7113ff64630aa282e45191255","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2023-05-05 14:06:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"15a42415de1b9f536045e92e5b137c52e380c74b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 14:07:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3246e08df64610453eace87137ec75eeeb41d043","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 16:12:33.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/01d8a4994a044797b48f980fa0d013a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70400aca5c754df2a801a5aef00e2f7a : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01168c09c31a4912aa3cd5e5517ba4f3 : SUCCESS in 38m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cffab65f2cfa499fbb666b19d5be478f : TIMED_OUT in 40m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afa70326f54e47fd80ecbfaa2c5f8f9c : SUCCESS in 13m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/bb408140eb7945ff8d8b35e786593622 : SUCCESS in 1h 23m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f5722980f1849fa806c97f8a7669f51 : TIMED_OUT in 2h 04m 30s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/2e06124ab0d04d4ab8ae4fe586218efe : SUCCESS in 39m 28s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/e7b70d727aef4a559592542d4e05f841 : SUCCESS in 41m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2284c15f92704676a9d435185d154337 : SUCCESS in 1h 00m 48s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/fd075aa7e06f4b03bb021d59c3c8d43d : SUCCESS in 53m 14s","accounts_in_message":[],"_revision_number":1},{"id":"b6390feaf3066d1ffda95927112b32b27338e24f","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-05-16 15:11:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"74826c6d9cd91d34dff59429b774e4c4cf277a1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 15:21:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ef4fb9bf18f469b8e60bd83ca8b3b19\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b3b901b08e64e409be363246000fa6c : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d97ea00e78a34ebb8e64a2876a04dfbe : SUCCESS in 9m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7fbcba2c49ba9e89d645c826537a4b80ffdbf467","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 16:24:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/148b70a5fe76452a9c273d0789b79540\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1953407f1e9f4d94abb2fdcc7e43021d : SUCCESS in 15m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c7c19ac11c94cf8b1c2867f11a4ffda : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30512452fd054193976e4f42864f544f : SUCCESS in 12m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d9d8f5cbd504efd9e5335b40b389494 : SUCCESS in 11m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d448470d2084f899ff122848e92ba0b : SUCCESS in 13m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cb2db07147247dfb4c624c5bae43e16 : SUCCESS in 11m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/0f4c0b776db8439ca6d0945dadca178d : SUCCESS in 48m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d75eeefba72042ce9097e2f8b9cf0109 : SUCCESS in 1h 11m 11s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/353de18965574fee82eefb3a30113707 : SUCCESS in 37m 17s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/564c93e6b8ed48058bb3b4738941c087 : SUCCESS in 45m 14s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f9360e501ab484fba92e50fe65af7f0 : SUCCESS in 33m 07s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/22037442f5764f44abbab23d4abf0be6 : SUCCESS in 37m 06s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/2a8f8ffe8dd24dc28ced9ababa72dbf9 : SUCCESS in 21m 18s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/6e73428b0c8344839af5b9eeb04043ea : FAILURE in 21m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6853bb35dbb4c90bf5f8c241eef9173 : SUCCESS in 59m 40s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/1b94674a04904d0295afcb3c4acaa315 : SUCCESS in 46m 55s","accounts_in_message":[],"_revision_number":1},{"id":"633df3dd3de4f058453879f4923c4c959f917800","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 16:24:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8fdf36d955edb0ae2f6a28df7e34bdad993e834","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 18:07:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fda35c3d41ad4c12ad61ca426d62a054\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cd115f01cde47d898aba24efd2162d7 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8df04b62e9b14f7cb0d8cb942c22c38f : SUCCESS in 11m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2504b40adff14fa29a75c192aa03517a : SUCCESS in 14m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0e07a5cc0234dd39517e77e0ea48a82 : SUCCESS in 16m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81695707748c45329607f1b37b3ad8e9 : SUCCESS in 13m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/3b3ead7ee06b42b59915e74d702aaeed : SUCCESS in 54m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d75f44c33e84b67b5fee5e75dd5adb2 : SUCCESS in 1h 29m 43s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/ce9a91192cc040ed9018d9951ccd2b6a : SUCCESS in 25m 44s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/74f4fa57a862487da3e99ca3c5b022a9 : SUCCESS in 27m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6d791b6f0574062986213be34f5c88d : SUCCESS in 43m 21s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/aa0d3cf8c2ae40d9bcdb12b0969f7cbe : SUCCESS in 50m 40s","accounts_in_message":[],"_revision_number":1},{"id":"bea71271cbc82da46fd48062f11f4ffff562654d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 18:07:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"384951a432f08f0cf033d4e241c705f3ae2ba1c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 18:08:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b2938ce15894f1d854e78a8e7b7011c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d36cf328b2c74728a320ed13d8f95571 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"42e2f985b246aa11de914bf5a8f0dc7f6836f501","revisions":{"42e2f985b246aa11de914bf5a8f0dc7f6836f501":{"kind":"REWORK","_number":1,"created":"2023-03-22 03:59:16.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/87/878187/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/87/878187/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/87/878187/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/87/878187/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/87/878187/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/87/878187/1"}}},"commit":{"parents":[{"commit":"f5db9801c23bde15d162a67d4fd6621e5bd09719","subject":"Remove Dependency on Cryptography \u003e\u003d36.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f5db9801c23bde15d162a67d4fd6621e5bd09719"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2023-03-22 03:56:58.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2023-03-22 03:56:58.000000000","tz":-300},"subject":"[PooledLDAPHandler] Clean up the fix for result3()","message":"[PooledLDAPHandler] Clean up the fix for result3()\n\nAn empty exception clause is unnecessary when you\u0027re using\na \"finally\" clause.\n\nPrevious-Change-Id: I59ebf0fa77391d49b2349e918fc55f96318c42a6\nChange-Id: I903db2fd2ac810ec96dbd25fc6529752c08f9a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/42e2f985b246aa11de914bf5a8f0dc7f6836f501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/42e2f985b246aa11de914bf5a8f0dc7f6836f501"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
