)]}'
{"id":"openstack%2Fkeystone~880456","triplet_id":"openstack%2Fkeystone~master~I8c45131b298ceae7b43b42e2c5df167607d18c48","project":"openstack/keystone","branch":"master","topic":"appcred-validation-speed2","attention_set":{},"removed_from_attention_set":{"21107":{"account":{"_account_id":21107,"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","username":"lathiat"},"last_update":"2024-01-26 17:37:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8c45131b298ceae7b43b42e2c5df167607d18c48","subject":"Improve application credential validation speed","status":"MERGED","created":"2023-04-14 07:31:22.000000000","updated":"2024-01-26 17:37:13.000000000","submitted":"2024-01-26 17:37:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"880456-appcred-validation-speed2","meta_rev_id":"5f1e0647d87f99fe7a5fa750dd4399262b14ed0d","_number":880456,"virtual_id_number":880456,"owner":{"_account_id":21107,"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","username":"lathiat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21107,"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","username":"lathiat"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-26 17:37:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"_account_id":21107,"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","username":"lathiat"},{"value":2,"date":"2024-01-26 15:33:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2024-01-26 15:25:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"_account_id":21107,"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","username":"lathiat"},{"value":1,"date":"2024-01-26 15:33:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":21107,"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","username":"lathiat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-14 07:43:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-14 09:02:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-26 15:25:45.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2024-01-26 15:33:26.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"7d874f3ab60d560d8788c2606bfbea6c597a941b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21107,"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","username":"lathiat"},"date":"2023-04-14 07:31:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b77d183296b4a71eeb30628bfc8bdefd82effbfa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 07:43:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a67ef565b3e54b62b05c86faca7cfafe\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fab7d408619c458dbd255811f2cb9ef7 : FAILURE in 8m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e56d5b57b35d46d38e3074d31b128c50 : FAILURE in 9m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a3b97a458f85a1f396a4a7a918edc6d1b6b88aca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 09:02:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9dfb333aeb57444bbbf789526c13a58f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0aa6a27b62445c3b62f4eb5468dc261 : FAILURE in 17m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a677d5ae9734409bbce5f37400981872 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a8aa6f25bda4c249d918d6c9c14f659 : FAILURE in 13m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6fd52e547b5644f7acf25906b45e49bd : FAILURE in 11m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b5f8070e41b47d1bb43c3da2c88574a : SUCCESS in 12m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/49f067452739471a9c6023d37cde3686 : SUCCESS in 1h 05m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f6bb0bec46b4a95822bd081732b85f8 : SUCCESS in 1h 25m 39s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/ba3d11fb0e48434c96bae02074b9cf9e : SUCCESS in 38m 22s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/c6aa5180ee7e42e187fb34a502a0be17 : SUCCESS in 33m 11s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6796045129d34a76bd5d0933fef8edd8 : SUCCESS in 38m 22s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/06da95a1cc1d45298fa093e17eeb7d56 : SUCCESS in 35m 16s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d448673bfacd4bf580a9351d1737504f : SUCCESS in 19m 00s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/88b63cb122cd412c885361892c0e389a : FAILURE in 23m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b8db773189740f68da1c9f95170e843 : SUCCESS in 54m 17s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/09d2118513ba4dbb967966261548c291 : SUCCESS in 41m 18s","accounts_in_message":[],"_revision_number":1},{"id":"6f0c6b1c86cb0882004f37bd09e400bfda881793","author":{"_account_id":21107,"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","username":"lathiat"},"date":"2023-04-16 07:13:30.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"07ec564c8d54d8e9f2cff225ace96536d9d6f35d","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":21107,"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","username":"lathiat"},"date":"2023-04-16 07:13:43.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"cf474715a5d60a8bfddef3c686a3128bd4b55fbe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21107,"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","username":"lathiat"},"date":"2023-11-29 02:15:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"06b8e51cb4cef2a4e6d48d0218a44a94a312b459","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 02:32:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/295e799fbf7e4a3c88785a9e619ab8b2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2f7ffb2d4f0e43b1875556da03910ffe : FAILURE in 15m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a75ed04ab5044c928641b27a3181f7d5 : FAILURE in 14m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b1e4801670f3c8698e41c93876dfd7fffa198f4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 04:12:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa4fe11c5e4c4f48927ab665bc0f228e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfe10d987ad44ef387f8736fa6885020 : FAILURE in 14m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f3805d835fd45a082337ccceaf22cea : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/458d8fdace674c4d97177a1a3e7d524d : FAILURE in 10m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ba080ac62e54bc49810a025fb071d0c : RETRY_LIMIT in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bf19d67a354428c93c659a814311bf2 : SUCCESS in 13m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/33dd15348e124c5493956310b3c09059 : SUCCESS in 56m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b700a2acf86f483e89e8cb9491bc92a6 : SUCCESS in 1h 36m 41s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c03242a833ec40d8a5217e91e6ea64a9 : FAILURE in 14m 11s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/dfb27ec9d9d34039a1e0df491dab61be : SUCCESS in 33m 31s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/822c806d60364f339215c1ca60a93562 : SUCCESS in 47m 38s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/078217d0a3ab4ed09636231000d7ca34 : SUCCESS in 34m 15s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/f57cddf184d54911b5de99336cf4fa7b : SUCCESS in 38m 01s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/e02f1e033d23443dad3579259f012abb : SUCCESS in 21m 16s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/cdc71a498f6540e4a89fcfb4e11ce084 : FAILURE in 26m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfffc45be5004bcfaa53a99966eaca74 : SUCCESS in 53m 18s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/062f094c4e3548e488b894dc480091ee : SUCCESS in 41m 30s","accounts_in_message":[],"_revision_number":2},{"id":"103337a346ee3a8d43397e4e3e22df72881b20ff","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21107,"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","username":"lathiat"},"date":"2024-01-02 03:41:07.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"39b35fda2fb808642d197dd6394966e14f8dc42f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-02 04:20:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3ae602a693a481d8bacf2c45a8e178e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c6ccb8cc52348f68f46712880f15165 : FAILURE in 10m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/abc4bbd4e5f345c1bcd380856668ce0c : FAILURE in 37m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"73a16f33fecb3bb35f0eb09a34229cd54b8a15a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-02 05:02:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/869f41d25855453dbe142050f97032a9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5cd20c388ad4c7bbf4102a455b719bc : FAILURE in 13m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ef940d6987e4fe084629745a882e3d9 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e010eca030974d07b19cb7352e17853f : FAILURE in 12m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8879162292fd4e4aa765711b52bb8711 : FAILURE in 11m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7e7506d7be740c0b20c4a78037500e1 : SUCCESS in 14m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/d6b14ac57a4449f9bddb45dbb27ec0c5 : SUCCESS in 46m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d44854c7ac7492d986da000fb34c753 : SUCCESS in 1h 16m 36s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8a065f96e2a14aa589c0e25f45640303 : FAILURE in 14m 26s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/7e8ef6d20aff434f890f29245abf85f8 : SUCCESS in 23m 01s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/cc53bc095e1a48f9addcc11e4d84e3d1 : SUCCESS in 48m 18s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7513eb685b84454bd520cb3a44f583d : FAILURE in 21m 50s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/a9ff95d2daf24a59bd2ad764aae74edf : SUCCESS in 20m 30s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/6bda76cc46624dc1a516c5b3dc326790 : SUCCESS in 22m 29s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/59218b8d6bde4e3a97b25e071b52e383 : FAILURE in 13m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37dbd598c6934ce48bf025af11678ccc : SUCCESS in 41m 15s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/b56bd1911316438b818bb45f3dc26bca : SUCCESS in 45m 36s","accounts_in_message":[],"_revision_number":3},{"id":"7c6c707a03431add8374b2965af2ea2fdc8ecf5a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21107,"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","username":"lathiat"},"date":"2024-01-02 08:50:45.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ffdcf8f5007730a13b0fcad7bf0c98589d94052d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":21107,"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","username":"lathiat"},"date":"2024-01-02 08:56:00.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"83115c84ad13583e9c0dbc8c347605e5379208e9","author":{"_account_id":21107,"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","username":"lathiat"},"date":"2024-01-02 08:57:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"016f787a87fd1c9c6a6ad83b1336e59326cd1359","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-02 09:25:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd267b3772b640fb8316dc43278b9205\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/14db091b4f0f430fbacab2354aae6a6b : SUCCESS in 33m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6b3bb6fa9f794f399093bbb85034c99c : SUCCESS in 15m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"498313b325c1c58594f79b641eaeea10b2a32b52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-02 10:26:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6787f1fd68642e599c752bce9289ead\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0f1a7ba35394217a46edbb3f719acae : SUCCESS in 14m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a24f672b6d6459cb91aea597a516f13 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6badf7b11ad74e5fabb7c5f233b7e1c9 : SUCCESS in 10m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ab7bd24a38bf47e0b2b7e84e60d6757a : SUCCESS in 10m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5527fcca1d148d2b7f9dd18a2b0e0b1 : SUCCESS in 12m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/23cd287f4c664c4e8682508f565dcaa7 : SUCCESS in 1h 00m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a48cf25fa0434229ba1674511c888651 : SUCCESS in 1h 30m 07s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/57569e600b0046e2a7e5e3fe72c878c5 : SUCCESS in 14m 16s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f4ee281ca83a4006a48e9d20894917e9 : SUCCESS in 31m 41s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/22549be84ee24ad4ac358e0ea87087ba : SUCCESS in 50m 06s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f78bfc1b4d5c44308a6f689356fe0fd0 : FAILURE in 21m 26s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/a0569b88b5944939b2fbc7306ab63394 : SUCCESS in 19m 44s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/2e01a2e2535a49bc81090d4da233b2d9 : SUCCESS in 9m 22s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/605865c0d282451a86e40e81b5de516e : FAILURE in 22m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a48477bcaa984048ac2d771446507bb7 : SUCCESS in 1h 04m 58s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/1cc1991945ef401f971b33d92d5710cc : SUCCESS in 43m 26s","accounts_in_message":[],"_revision_number":4},{"id":"d89a38674e9814fac40091707a1ba504c0208651","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-01-26 15:25:45.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"63c82f025d3e9f603a1d1fe0bbbfed7ce4530a8a","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-01-26 15:33:26.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6b7c815ddb9a2cbd76f9b2a5285e97c43df5fd2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 15:34:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ef611f8df743a5235ef76e0ccfe8038994af9d3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 17:37:13.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23add62a99c64ce297dfd78e65352797\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b2f3d7a4e2a4d6f925c268787a7ade6 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8309696aa89d4db4ae36bce5fe3af83b : SUCCESS in 10m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa2955bcb9034a63b74eec9af1f6bf33 : SUCCESS in 13m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce66359982134b9ca3608a430f8ab5ff : SUCCESS in 10m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/e5dd79f2bd86491f9501535633956509 : SUCCESS in 57m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7165178d462340069fd6e81d1a1f9886 : SUCCESS in 1h 40m 09s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ad6698efe2da4e19951286005a7c0413 : SUCCESS in 13m 30s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/32590c5a7e1245a98430ca51ad11c925 : SUCCESS in 28m 02s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/85ff19218d0b4bd698feb0b45a071e93 : SUCCESS in 35m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6cfafdb4cf184ae9a35607d3d4eec104 : SUCCESS in 48m 36s","accounts_in_message":[],"_revision_number":4},{"id":"5f1e0647d87f99fe7a5fa750dd4399262b14ed0d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 17:37:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"67b5cca032c36ef67a8aee9db6adada9b403dc5c","revisions":{"013f8a2bb20b409ea2fece74dde392765b6d49a0":{"kind":"REWORK","_number":1,"created":"2023-04-14 07:31:22.000000000","uploader":{"_account_id":21107,"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","username":"lathiat"},"ref":"refs/changes/56/880456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/56/880456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/56/880456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/56/880456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/56/880456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/56/880456/1"}}},"commit":{"parents":[{"commit":"f5db9801c23bde15d162a67d4fd6621e5bd09719","subject":"Remove Dependency on Cryptography \u003e\u003d36.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f5db9801c23bde15d162a67d4fd6621e5bd09719"}]}],"author":{"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","date":"2023-04-14 07:15:15.000000000","tz":480},"committer":{"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","date":"2023-04-14 07:31:16.000000000","tz":480},"subject":"Improve application credential validation speed","message":"Improve application credential validation speed\n\nValidating an application credential token is very slow, taking at least\n400ms+ in a simple devstack environment, 5-10x longer than validating a\nuser/password project token.\n\nThe primary bottleneck during a token validation request\n(/v3/auth/tokens) is that token.roles is evaluated at least 5 times.\nvalidate_token is called twice, first during RBAC to populate the\nsubject token context and again to actually validate the token. Each\ncall to validate_token then called token.roles twice because it first\nchecks if it is None, before calling it again to use the result. Lastly\ntoken.roles is evaluated a fifth time during\nrender_token_response_from_model.\n\nEach evaluation of token.roles calls through\n_get_application_credential_roles into list_role_assignments which then\nmakes multiple round-trip SQL queries to the database.\n\nUnlike the related get_roles_for_user_and_project function, none of\nthese calls are currently cached/memoized. We memoize\nlist_role_assignments to get the same-speedup.\n\nReduce the number of token.roles calls to only 3 by storing and re-using\nthe token.roles result in validate_token, then memoize\nlist_role_assignments so the 2nd and 3rd call fetch from the cache\ninstead of repeating many SQL queries.\n\nThis provides a substantial performance improvement bringing validation\ntime in-line with user/password tokens.\n\nne\n\nChange-Id: I8c45131b298ceae7b43b42e2c5df167607d18c48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/013f8a2bb20b409ea2fece74dde392765b6d49a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/013f8a2bb20b409ea2fece74dde392765b6d49a0"}]},"branch":"refs/heads/master"},"5a9dcc3eafe6eaf1a0c0f8a39f517b0c914d6026":{"kind":"REWORK","_number":2,"created":"2023-11-29 02:15:18.000000000","uploader":{"_account_id":21107,"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","username":"lathiat"},"ref":"refs/changes/56/880456/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/56/880456/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/56/880456/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/56/880456/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/56/880456/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/56/880456/2"}}},"commit":{"parents":[{"commit":"fe1a75cf3a0ed7be19d8e5a2fa99e5b2947cfa10","subject":"Merge \"doc: Update the installtion guide for RHEL8/CentOS8 and RHEL9/CentOS9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/fe1a75cf3a0ed7be19d8e5a2fa99e5b2947cfa10"}]}],"author":{"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","date":"2023-11-29 02:14:51.000000000","tz":0},"committer":{"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","date":"2023-11-29 02:14:51.000000000","tz":0},"subject":"Improve application credential validation speed","message":"Improve application credential validation speed\n\nValidating an application credential token is very slow, taking at least\n400ms+ in a simple devstack environment, 5-10x longer than validating a\nuser/password project token.\n\nThe primary bottleneck during a token validation request\n(/v3/auth/tokens) is that token.roles is evaluated at least 5 times.\nvalidate_token is called twice, first during RBAC to populate the\nsubject token context and again to actually validate the token. Each\ncall to validate_token then called token.roles twice because it first\nchecks if it is None, before calling it again to use the result. Lastly\ntoken.roles is evaluated a fifth time during\nrender_token_response_from_model.\n\nEach evaluation of token.roles calls through\n_get_application_credential_roles into list_role_assignments which then\nmakes multiple round-trip SQL queries to the database.\n\nUnlike the related get_roles_for_user_and_project function, none of\nthese calls are currently cached/memoized. We memoize\nlist_role_assignments to get the same-speedup.\n\nReduce the number of token.roles calls to only 3 by storing and re-using\nthe token.roles result in validate_token, then memoize\nlist_role_assignments so the 2nd and 3rd call fetch from the cache\ninstead of repeating many SQL queries.\n\nThis provides a substantial performance improvement bringing validation\ntime in-line with user/password tokens.\n\nChange-Id: I8c45131b298ceae7b43b42e2c5df167607d18c48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5a9dcc3eafe6eaf1a0c0f8a39f517b0c914d6026"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5a9dcc3eafe6eaf1a0c0f8a39f517b0c914d6026"}]},"branch":"refs/heads/master"},"ed75c5757d2ca30d2df492042518dd50ec0d8971":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-01-02 03:41:07.000000000","uploader":{"_account_id":21107,"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","username":"lathiat"},"ref":"refs/changes/56/880456/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/56/880456/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/56/880456/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/56/880456/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/56/880456/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/56/880456/3"}}},"commit":{"parents":[{"commit":"9e4a3157dd326801b9ae0ef2b7934c982259b3dd","subject":"Merge \"Fix typo in cmd/status.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9e4a3157dd326801b9ae0ef2b7934c982259b3dd"}]}],"author":{"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","date":"2023-11-29 02:14:51.000000000","tz":0},"committer":{"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","date":"2024-01-02 03:39:48.000000000","tz":0},"subject":"Improve application credential validation speed","message":"Improve application credential validation speed\n\nValidating an application credential token is very slow, taking at least\n400ms+ in a simple devstack environment, 5-10x longer than validating a\nuser/password project token.\n\nThe primary bottleneck during a token validation request\n(/v3/auth/tokens) is that token.roles is evaluated at least 5 times.\nvalidate_token is called twice, first during RBAC to populate the\nsubject token context and again to actually validate the token. Each\ncall to validate_token then called token.roles twice because it first\nchecks if it is None, before calling it again to use the result. Lastly\ntoken.roles is evaluated a fifth time during\nrender_token_response_from_model.\n\nEach evaluation of token.roles calls through\n_get_application_credential_roles into list_role_assignments which then\nmakes multiple round-trip SQL queries to the database.\n\nUnlike the related get_roles_for_user_and_project function, none of\nthese calls are currently cached/memoized. We memoize\nlist_role_assignments to get the same-speedup.\n\nReduce the number of token.roles calls to only 3 by storing and re-using\nthe token.roles result in validate_token, then memoize\nlist_role_assignments so the 2nd and 3rd call fetch from the cache\ninstead of repeating many SQL queries.\n\nThis provides a substantial performance improvement bringing validation\ntime in-line with user/password tokens.\n\nChange-Id: I8c45131b298ceae7b43b42e2c5df167607d18c48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ed75c5757d2ca30d2df492042518dd50ec0d8971"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ed75c5757d2ca30d2df492042518dd50ec0d8971"}]},"branch":"refs/heads/master"},"67b5cca032c36ef67a8aee9db6adada9b403dc5c":{"kind":"REWORK","_number":4,"created":"2024-01-02 08:50:45.000000000","uploader":{"_account_id":21107,"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","username":"lathiat"},"ref":"refs/changes/56/880456/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/56/880456/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/56/880456/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/56/880456/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/56/880456/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/56/880456/4"}}},"commit":{"parents":[{"commit":"9e4a3157dd326801b9ae0ef2b7934c982259b3dd","subject":"Merge \"Fix typo in cmd/status.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9e4a3157dd326801b9ae0ef2b7934c982259b3dd"}]}],"author":{"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","date":"2024-01-02 08:46:24.000000000","tz":0},"committer":{"name":"Trent Lloyd","email":"trent.lloyd@canonical.com","date":"2024-01-02 08:46:24.000000000","tz":0},"subject":"Improve application credential validation speed","message":"Improve application credential validation speed\n\nValidating an application credential token is very slow, taking at least\n400ms+ in a simple devstack environment, 5-10x longer than validating a\nuser/password project token.\n\nThe primary bottleneck during a token validation request\n(/v3/auth/tokens) is that token.roles is evaluated at least 5 times.\nvalidate_token is called twice, first during RBAC to populate the\nsubject token context and again to actually validate the token. Each\ncall to validate_token then called token.roles twice because it first\nchecks if it is None, before calling it again to use the result. Lastly\ntoken.roles is evaluated a fifth time during\nrender_token_response_from_model.\n\nEach evaluation of token.roles calls through\n_get_application_credential_roles into list_role_assignments which then\nmakes multiple round-trip SQL queries to the database.\n\nUnlike the related get_roles_for_user_and_project function, none of\nthese calls are currently cached/memoized. We memoize\nlist_role_assignments to get the same-speedup.\n\nReduce the number of token.roles calls to only 3 by storing and re-using\nthe token.roles result in validate_token, then memoize\nlist_role_assignments so the 2nd and 3rd call fetch from the cache\ninstead of repeating many SQL queries.\n\nThis provides a substantial performance improvement bringing validation\ntime in-line with user/password tokens.\n\nChange-Id: I8c45131b298ceae7b43b42e2c5df167607d18c48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/67b5cca032c36ef67a8aee9db6adada9b403dc5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/67b5cca032c36ef67a8aee9db6adada9b403dc5c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
