)]}'
{"id":"openstack%2Fkeystone~882692","triplet_id":"openstack%2Fkeystone~master~I147f3888c42e2d8d25a0ddd20f4e3974c8a38632","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"35705":{"account":{"_account_id":35705,"name":"Arvid Requate","email":"requate@univention.de","username":"reqa"},"last_update":"2023-08-10 16:13:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I147f3888c42e2d8d25a0ddd20f4e3974c8a38632","subject":"Fix typo in openid federation diagram","status":"MERGED","created":"2023-05-09 11:10:04.000000000","updated":"2023-08-10 16:14:52.000000000","submitted":"2023-08-10 16:13:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882692","meta_rev_id":"f39658dc2ae76aca8725806228c348731ecc4972","_number":882692,"virtual_id_number":882692,"owner":{"_account_id":35705,"name":"Arvid Requate","email":"requate@univention.de","username":"reqa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-10 16:13:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2023-08-10 15:59:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2023-08-10 15:59:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-09 11:22:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-10 15:59:00.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"}],"messages":[{"id":"a6280b393c535b9385c27e42db4bc02ea8831760","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35705,"name":"Arvid Requate","email":"requate@univention.de","username":"reqa"},"date":"2023-05-09 11:10:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"354ba007e86af7501d449e88526b86138adbfbbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 11:22:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f113f36d65d640fe91241c6948ec8c01\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b4466b2fa5940a689e77010066d5085 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/698f11dfb3e44697b062cdbc03cbb3a6 : SUCCESS in 10m 49s","accounts_in_message":[],"_revision_number":1},{"id":"aa0c78a7fb01040d74800f2a5bd1b676ef3aaf62","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2023-08-10 15:59:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"836fadcedec7415ee8d2c4b3781f356eb786ed75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 15:59:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c02c8b66dc2bc01a5a2241a53489efc87af93adf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 16:13:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cdf4224fb3348238c88660c02626316\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e94dfd07ea846bc8c95beefed61feec : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e20ee445fc94cefad1872131f15f6b0 : SUCCESS in 12m 55s","accounts_in_message":[],"_revision_number":1},{"id":"922e1214bdfb06fcef260bfdcfc07148a73b4526","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 16:13:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f39658dc2ae76aca8725806228c348731ecc4972","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 16:14:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/411da90e6b5a4d7786bfe34adf96c982\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd605bd047a24a9980d6d2afac0a523b : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1195c38b8b04a850a2d44f44f19d19fe0b5e2765","revisions":{"1195c38b8b04a850a2d44f44f19d19fe0b5e2765":{"kind":"REWORK","_number":1,"created":"2023-05-09 11:10:04.000000000","uploader":{"_account_id":35705,"name":"Arvid Requate","email":"requate@univention.de","username":"reqa"},"ref":"refs/changes/92/882692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/92/882692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/92/882692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/92/882692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/92/882692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/92/882692/1"}}},"commit":{"parents":[{"commit":"1d58835d3e388363af9458bb4a145a9472a22e7b","subject":"Merge \"Print a human readable error if tls certs are not provided\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1d58835d3e388363af9458bb4a145a9472a22e7b"}]}],"author":{"name":"Arvid Requate","email":"requate@univention.de","date":"2023-05-09 11:08:11.000000000","tz":120},"committer":{"name":"Arvid Requate","email":"requate@univention.de","date":"2023-05-09 11:08:11.000000000","tz":120},"subject":"Fix typo in openid federation diagram","message":"Fix typo in openid federation diagram\n\nSection \"OpenID Connect with keystone and horizon\"\nshould use openid endpoint instead of saml2.\n\nChange-Id: I147f3888c42e2d8d25a0ddd20f4e3974c8a38632\nSigned-off-by: Arvid Requate \u003crequate@univention.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1195c38b8b04a850a2d44f44f19d19fe0b5e2765"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1195c38b8b04a850a2d44f44f19d19fe0b5e2765"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
