)]}'
{"id":"openstack%2Fkeystone~886509","triplet_id":"openstack%2Fkeystone~master~Id92d74d368356d67d5a9be6d3eada44cd190a35d","project":"openstack/keystone","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-07-05 20:40:42.000000000","reason":"Change was submitted"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2023-07-05 18:14:17.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2023-07-04 08:37:47.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}},"hashtags":[],"change_id":"Id92d74d368356d67d5a9be6d3eada44cd190a35d","subject":"Add job to test with SQLAlchemy master (2.x)","status":"MERGED","created":"2023-06-20 15:33:40.000000000","updated":"2023-07-05 20:43:03.000000000","submitted":"2023-07-05 20:40:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886509-sqlalchemy-20","meta_rev_id":"a29079da7e1385bb97f3ff2c1bf867a038494c80","_number":886509,"virtual_id_number":886509,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-05 20:40:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2023-07-05 18:14:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2023-07-04 08:37:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2023-07-05 18:14:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-20 15:38:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-20 18:55:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-03 10:29:30.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2023-07-03 10:29:30.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"911a7c0f6e7e172a8263d079d9a45626be7fb423","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-06-20 15:33:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17374b791c7de93882b0593b1ec056b0e57fda1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 15:38:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b25e2d337d94f2da79e03cb11c04829\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/60d9a63f672c487590d7cfcf474786ad : NODE_FAILURE Node request 300-0021496946 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d8fa07f2e32f4c38b6afaa67c7990cc3 : NODE_FAILURE Node request 300-0021496947 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e0e09afe1ad1c489af56366f028dececca556c4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 18:55:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf1580ce5b654339ab15eb8de693d82d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7834f5501684d859c5bcb5506f9e53a : SUCCESS in 17m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15c543c4b738402a9b1158dfc2198b26 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d86b9e37e0d64c88aa40fe6a5149283e : SUCCESS in 14m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d443410cea49477bbeeb4f6938f91975 : SUCCESS in 15m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84c6e4d62eec42a89c0f01cfce238955 : SUCCESS in 12m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aad12b87cb1141bc8794d575772d0970 : SUCCESS in 11m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/ee28fcdd746244c9b18e229908493628 : SUCCESS in 54m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8772a34390d84afa80907e71bf9b6288 : SUCCESS in 1h 43m 05s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/df50f6bde5284b48b182510956574fd5 : FAILURE in 16m 30s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/7032870fc3f047d3b553301d67c7803c : SUCCESS in 36m 24s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/110e0f9599244a0ab00909f4c48e0d60 : SUCCESS in 50m 03s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acbffb8775bd4d95b88a0c0f8509c78d : FAILURE in 35m 38s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/c39e407de9c047c8b4c84a0bfa17fda5 : SUCCESS in 23m 36s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/6a9978ec0ed247668d120fb59099387f : SUCCESS in 19m 20s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/731d280bea4e4d638e335fa76c7b35c0 : FAILURE in 28m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ea292c529ce472992fc92353ce484e0 : SUCCESS in 51m 37s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/51ffbbeab7fa49a4ade7ef9f8b765d84 : SUCCESS in 37m 50s","accounts_in_message":[],"_revision_number":1},{"id":"e036db723e66e47bdb3d1844a0afea26ac508ad6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-06-27 09:58:52.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"af9dbea74f547c99848e33abb7b21c02a6cf0ac1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 10:19:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bc98599a887432087c66905871c2f42\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3334af8b6a0445f1b4700e3757c5c29b : SUCCESS in 13m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/31b1ad3b8f9d41139d21ba05cd516dae : SUCCESS in 13m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a6fb80355db03a8ce34c9d39b75fdfcb29a0c95a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 12:21:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32a8e85aae2b4ebf81ce504b4f9e9576\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab01a40238844668a97061a58662fda5 : SUCCESS in 19m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8b320bd43a64a5d9f079dbd02209860 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94329a75c3944731a0688df0d4c99b1f : SUCCESS in 10m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0243fcc940584138b25b06ba2308c27c : SUCCESS in 11m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05f5d8dc4da24aeaa6b6028d9d254b62 : SUCCESS in 12m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61555276a0c94748ab6fec522bd45b24 : SUCCESS in 14m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/909171f979ec4369ad2d413ae816ad8e : SUCCESS in 1h 09m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4c513482d8b44faa9c52e6a58fab39d : SUCCESS in 1h 48m 33s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c95e4396ee3440e5a06ec9d398d80041 : SUCCESS in 21m 34s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d421c754f6e340d491d923cecbaca76a : SUCCESS in 36m 27s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/b93d6d410c6f4930812497f1573eb1da : SUCCESS in 32m 06s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8993fa356d524bf596be0ae24752145c : FAILURE in 42m 34s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/47a15b8a4de447d886e913eca349d186 : SUCCESS in 47m 13s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/cd66e561aa48427c9fab477f4616c3fd : SUCCESS in 11m 01s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/3ac73229bc474cbc99c28962e40a9c77 : FAILURE in 13m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bba243339974a22989b70a793546c55 : SUCCESS in 1h 07m 59s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/99fd1adb794548579768bd3f519b742b : SUCCESS in 58m 55s","accounts_in_message":[],"_revision_number":2},{"id":"754d7ddddf93526a1769bc826fd3467cdc8217bf","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-07-04 08:37:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dcb34b60d07a88b4df65b93fa0e0da83955e01ef","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-07-05 18:14:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ba65da529ae858198b8a5a7424dc837b3b2adf3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 18:14:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"00d8fefb15398821bfaaf54087f771fc00c2973a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 20:40:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa0eb06d551343e6aadab87b37a923fe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2161328eb6f14c569b1c43755df2de23 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/183338d72fff496f925b28db0d7b6780 : SUCCESS in 12m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e89362f285524fada20bd92d67adb3ad : SUCCESS in 14m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/369168d21eb244b89724895c24847b90 : SUCCESS in 14m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/989dd97233914a419af167eb08fe9b66 : SUCCESS in 12m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/6e613f23e2084e37a8bfdac97409a7f1 : SUCCESS in 1h 03m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/085c0a6cff214fe7b66664b8523a8875 : SUCCESS in 1h 33m 57s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/325b23d0b2d24b80a1b896e73cf972c2 : SUCCESS in 12m 39s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d2c6afa3c6a64dadb7e0bdc5d542b9e2 : SUCCESS in 24m 37s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/b46a73cbc70e4e898b8cf50a5c97a986 : SUCCESS in 18m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c60e2c2f8754fcf85da56dbaea55ac4 : SUCCESS in 48m 27s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/0c80aee271bd4f5790091192ef120570 : SUCCESS in 37m 32s","accounts_in_message":[],"_revision_number":2},{"id":"5d74d0db6ae39e4b561748cb6af07444f4f05b03","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 20:40:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a29079da7e1385bb97f3ff2c1bf867a038494c80","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 20:43:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7588302314284aa5bc50c7e459602a3a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73b3fca2e757453e824dbd5081e4f6fc : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b2d638e902b0164b1a9531aa1d12a56f103206ec","revisions":{"2fd52713bb64b6c68c670a98cc4e90cc106ce9be":{"kind":"REWORK","_number":1,"created":"2023-06-20 15:33:40.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/09/886509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/09/886509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/886509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/886509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/886509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/09/886509/1"}}},"commit":{"parents":[{"commit":"535bc8e22ecb67eaa17a7a0a61037aac6f602b21","subject":"sql: Remove duplicate constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/535bc8e22ecb67eaa17a7a0a61037aac6f602b21"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-06-20 15:32:43.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-06-20 15:32:43.000000000","tz":60},"subject":"Add job to test with SQLAlchemy master (2.x)","message":"Add job to test with SQLAlchemy master (2.x)\n\nChange-Id: Id92d74d368356d67d5a9be6d3eada44cd190a35d\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2fd52713bb64b6c68c670a98cc4e90cc106ce9be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2fd52713bb64b6c68c670a98cc4e90cc106ce9be"}]},"branch":"refs/heads/master"},"b2d638e902b0164b1a9531aa1d12a56f103206ec":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-27 09:58:52.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/09/886509/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/09/886509/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/886509/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/886509/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/886509/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/09/886509/2"}}},"commit":{"parents":[{"commit":"f477307bd8e71f0d9c37e67e79738fa36359dfbc","subject":"db: Don\u0027t rely on branched connections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f477307bd8e71f0d9c37e67e79738fa36359dfbc"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-06-20 15:32:43.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-06-27 09:56:50.000000000","tz":60},"subject":"Add job to test with SQLAlchemy master (2.x)","message":"Add job to test with SQLAlchemy master (2.x)\n\nChange-Id: Id92d74d368356d67d5a9be6d3eada44cd190a35d\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b2d638e902b0164b1a9531aa1d12a56f103206ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b2d638e902b0164b1a9531aa1d12a56f103206ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
