)]}'
{"id":"openstack%2Fkeystone~887509","triplet_id":"openstack%2Fkeystone~master~I74e47a9ed986c7c3af19676ac65f4d290bcb4cc0","project":"openstack/keystone","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-07-11 18:35:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I74e47a9ed986c7c3af19676ac65f4d290bcb4cc0","subject":"sql: Remove service_provider.relay_state_prefix default","status":"MERGED","created":"2023-07-03 11:12:31.000000000","updated":"2023-07-11 18:37:57.000000000","submitted":"2023-07-11 18:35:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887509-sqlalchemy-20","meta_rev_id":"d23b0097f720d94c959ef1def0a5e84e1c412345","_number":887509,"virtual_id_number":887509,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-11 18:35:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2023-07-06 15:51:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2023-07-07 14:15:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2023-07-07 14:15:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-03 11:29:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-03 13:36:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-06 15:51:04.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2023-07-07 14:15:27.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"91ac5acd26410e3264cd86a2242110f1986e3ec9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-03 11:12:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"237ad0793c44be1ff0746b0d637de15aa08e2cec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 11:29:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b6c1bfda3644efda5e0411f67010630\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee5efcee307246aeb791dea7c7e2625f : FAILURE in 13m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/defe273df7834ec78e37b1ed6ba766c2 : FAILURE in 11m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c9fe39febd026b149fe1d0474425325749a7fa8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-03 11:33:09.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"34f280b95c1808b3c796e28d8c055fa355d1713b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 11:58:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48a7940c2ba847909e90e4bbcb0efa3f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c3a6e1bbb5740559e3a6daab45dc475 : FAILURE in 21m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b19c0ab1ef5e476bb2f9d084c8b2f830 : FAILURE in 15m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b21a2fa0cb17a08fdc4bc1217b5e43efad5d63f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 13:36:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ed4fa8184b746bdbf6225dc9b46518a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bba28080810a40d0a0ac99dc02a1d41d : FAILURE in 13m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae45d6cb7fe1474298645eb90cc38859 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10a67ca6c2ba42179e673111297dac2f : FAILURE in 12m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72bb164e746746e4af85692137bd1046 : FAILURE in 13m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3af2551efe5b4a5291a450cbf23f7a66 : FAILURE in 13m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e8de395dbb34ce699ead4e6cad0e3fe : SUCCESS in 13m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/c393a89182504009bff45f8d07829139 : SUCCESS in 49m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e3649ca624c433d8ecf5d097a9d9571 : SUCCESS in 1h 47m 37s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/acf9815cd100419dbfdc78e16721d209 : FAILURE in 21m 59s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/2661d7a1d56547deb039731b86766f1c : SUCCESS in 37m 12s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/274c8eec30704eb2974379860f895080 : SUCCESS in 55m 30s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/543d3097b7fd41f4b465da9bd0799a2b : SUCCESS in 27m 30s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/1c1c89c9d24445818d75843b3e95ff3c : SUCCESS in 26m 03s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/fc394144d7944c96a7cb6c061433c907 : SUCCESS in 10m 31s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/99b8175dd5264f369a89711f8cd5be6d : FAILURE in 12m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aeb02c8036154cdf87cf6b3527a3eaee : SUCCESS in 46m 13s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/abbe4c0c2af749cc9fc2c63071d23b6b : SUCCESS in 37m 34s","accounts_in_message":[],"_revision_number":2},{"id":"f543d373ef41f9458cdd77e2829e1b38415660da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-06 09:23:32.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"2c99346d0c46297ba748b0b0ed105e54cca3b979","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 09:38:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea92397933ee4a3680c2ac37a64ac6f3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b96a37628bd445688a5f1c3c8caea83e : SUCCESS in 11m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b93cf8941e724d828edc7514a8388b77 : SUCCESS in 11m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"504bbf817299f515c5eaa2c0c0672d6313ee598a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 10:57:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1967ee4102c04a348331a47cf02f62b9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00a9f38e46d64448b6fc472aae39b822 : SUCCESS in 14m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f09253971ad4e96bd308d337d9eb608 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/227fc52c27a64b5c93f385a3509ca875 : SUCCESS in 11m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a2f9f34a2054e05ace5527a0a5b8bd2 : SUCCESS in 12m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6dacda2c2b334279804d9d81af11ead9 : SUCCESS in 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f73f4347b2e417bade5e3a4b8009a2f : SUCCESS in 11m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/8c943df69ec24b2fae468b255c019e90 : SUCCESS in 57m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1ad956e936e423c888b64dcee732587 : SUCCESS in 1h 31m 43s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ece0f13e6bfb449282dcf52bc6832220 : SUCCESS in 15m 57s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f3caf1107be0400eb5c6fad20a0bd9d7 : SUCCESS in 34m 47s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/434fe6bb7fe349db985d3f2a8c9ea8f2 : SUCCESS in 44m 13s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d6a4efc2fd240048179faafd2e4d8d4 : SUCCESS in 22m 47s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/8cde7d8cd59e4fe9abf1cbbc9b86914e : SUCCESS in 55m 08s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/9585f5d770a84784a4622875b28747b8 : SUCCESS in 20m 53s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/be93c67ed76341c7bd721ea7455c75c8 : FAILURE in 13m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c72a8b7f8fc942a2aeaf11c5c9b4b6a2 : SUCCESS in 1h 12m 12s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/b50c108e87984326883c2c4a4e6da59d : SUCCESS in 48m 58s","accounts_in_message":[],"_revision_number":3},{"id":"50a4960b206b55b535bab852d3f42dc43a2c3890","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-07-06 15:51:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"97b4ffdaa9367637216665333cc568e1dabb34c8","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-07-07 14:15:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"52df141e458f71c266b37ea9cbb82533955084cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 14:16:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"123eaca878f456ca6294adccefa4808c5cef0004","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 16:31:40.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0deabe333f0646ec8d2fc50d424385b5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10b245e141844787934faf38cbc47ca6 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eff25ea87a5a41a3a7d40f5612d223f2 : SUCCESS in 12m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ca48359f3624336a8cf4edd035b68ea : SUCCESS in 12m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f14c10b8437d459d9886066c6b62c75d : FAILURE in 11m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b60464d7e9f84a19b802ab649acb52a8 : SUCCESS in 12m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/bf29e8578c914ac5802fcec4e07e9efd : SUCCESS in 1h 01m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aae30377dfb44bf9acef5ef9eb1c4479 : FAILURE in 1h 10m 09s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/99ba5059b6e14d50a33337004a26c562 : SUCCESS in 18m 13s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/02ab436f47924c60993348730f750e74 : SUCCESS in 39m 17s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/156c1a6d881c4a20b54caf005ecf0413 : SUCCESS in 31m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0a759c152dd4ffb87cb908987c1e4e5 : SUCCESS in 1h 13m 20s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/08ebfbf7c13947a5bdbcd8b47776499b : SUCCESS in 41m 12s","accounts_in_message":[],"_revision_number":3},{"id":"14bacc8a13348e970a6e9612f4a6b0299e9bba3e","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-07-10 10:31:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4b0db1f4c13c0845e0572b1e490a6f61eb490a3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 10:48:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c7b84889f3742c290229de3e5703c7e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/343c479557db4aafb6fe785ba72a850c : SUCCESS in 10m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2d59503a1dfa419e8f4bc3a2049bde5c : SUCCESS in 11m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"553c5185b551fab67ce2a4ef01d5fb09751cacc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 12:57:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4642c91b0750419d98c54882f82a4fce\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfebd48d71b74ca4839c81edcced991f : SUCCESS in 13m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfd4c51727c348ada83e6f119c32165d : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d21f5b03b894253b646ac6662e2b26a : SUCCESS in 13m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8cb1aa875fb4093a97023483458ad0d : SUCCESS in 10m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee993aa2376b41db85a443ff9c88cd7e : SUCCESS in 11m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11427c0a141247bc8fea1a2446165f39 : SUCCESS in 11m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/9e7fadf489b047d9ac84e4a256ae0c0c : SUCCESS in 41m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1b64ec1719a741cd8dc0fd096c083332 : TIMED_OUT in 2h 21m 06s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8d6d29db8c4245c991feedde64880360 : SUCCESS in 12m 20s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/1a593d62c0244d98b7704dc1084851b5 : SUCCESS in 20m 21s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/fc092b94ea7d4743b74edb1064a5d4a2 : SUCCESS in 29m 15s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9f093062c17430dbe422689af6aed6a : SUCCESS in 21m 01s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/308d9fd7d1d442618aed695e4e3ad2b6 : SUCCESS in 22m 19s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/50ed7e4eba394aea9baf528eb042bdee : SUCCESS in 21m 22s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/13020724dd4740e7863843271bf17f5b : FAILURE in 9m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79e6dda949744f84bab0625804a2acc2 : SUCCESS in 1h 03m 11s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/8ccbc26cca6e4ccead022fae115b3ccb : SUCCESS in 47m 52s","accounts_in_message":[],"_revision_number":3},{"id":"1d9003a33312e683179d9d83af33251a43cf75c7","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-07-10 13:12:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8edf419662d19f90f15b20cdcfe3b835b6d899cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 13:27:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efb8a8e49d944fd4b299a794ea47a760\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/93cfb9a472b84b1db526e493884a2afe : SUCCESS in 12m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cfc047caee5949e4ad38e1f8e7cc4d7f : SUCCESS in 12m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"486429e58526dae8a5f3d75300062d2bf263b487","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 15:31:17.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d41ad655100147f6844ef15d43c6a0c6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07e49db2f92f4ee7a88353fa6fd2f580 : SUCCESS in 13m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/568f6aecf36247ea9e0bc4c0bee5316b : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3811d5ed1044302b29d20610560798f : SUCCESS in 9m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b67e30c5c4164c7cadc00913f4168b2a : SUCCESS in 14m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90ea089657944f1389bcb19b5f1b5e97 : SUCCESS in 11m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f09931eb5be348738820f40d1896532f : SUCCESS in 15m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/f83f2d46259546b4bc5fd60d56c9077d : SUCCESS in 44m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f4d99a5280bd4976a0b69d287f262cca : TIMED_OUT in 2h 04m 48s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/71015726b6b74414be56d1cb57cde038 : SUCCESS in 15m 44s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/b097ba1930784bb6bbcc014cce72f1cc : SUCCESS in 23m 33s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/25094ebd2d714ec0bfdb1b61f9900c32 : SUCCESS in 54m 47s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/847000e2c4a54501a8aa08157e534b9e : SUCCESS in 22m 03s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/5edeed26462e40f28479c852d0b9766c : SUCCESS in 43m 03s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/7032d0815875435cbd514e47cb5d551b : SUCCESS in 24m 46s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/86b59f7c10c741a0a7ec7de8d9731574 : FAILURE in 11m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8666446af6734b8b9afd975ba2106da9 : SUCCESS in 1h 10m 51s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/cc963a74e9674a4a90a942eae2216a44 : SUCCESS in 57m 05s","accounts_in_message":[],"_revision_number":3},{"id":"9ffd2b8673f8fd8c2af35e1a4c8f5ecc61051276","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-07-11 12:40:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6e37dad0d46bf671952dbbcb4921da51246c7049","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 13:23:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7a962c3e6304c289327bca092e0e254\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a1558e08f8c470d8b94476c680dda37 : SUCCESS in 40m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2dd8d507344140079cfe8857310d4dbc : SUCCESS in 10m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"036b14477f761b120571083acdbaf5238dfa4b48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 14:30:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d9411525150454b9e453d9caedfb28e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7e4c0ae615b49c78d1d89eddf0c7590 : SUCCESS in 15m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b8bf427317e47f39a46daff8fe0a107 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/696cb12677fe4d699c7cb99b0b8da3e3 : SUCCESS in 13m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/914d1514e4e141aabe0b2fccf2e44184 : SUCCESS in 14m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d6509c9d2874af695e15c83480121fd : SUCCESS in 10m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aec87126731d4c4bbdf1cb05d619a92c : SUCCESS in 11m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/26d6f1257b8e4bf4a5924a238ecfa394 : SUCCESS in 1h 09m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0cc7a7eb6c624580a5530ff6a3fad050 : SUCCESS in 1h 47m 31s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e68d2e0d9b60434fa4420a44379efe40 : SUCCESS in 12m 18s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/2e2aac3941b1439e9ff384e21829da1b : SUCCESS in 35m 43s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/5bccf766f9a44564bb011675814ddb22 : SUCCESS in 32m 42s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75a78f1a1173404a9e9cc12ff602c369 : SUCCESS in 26m 32s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/4e26145b1cca45e89162e2ddfc7f8a16 : SUCCESS in 22m 29s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/9e9e8663691d41d2820805afd78f7dcd : SUCCESS in 9m 59s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/0aa651c206e842ae8a82551d8e3eef24 : FAILURE in 12m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3abcea6a7a554de19a5e5da48def3297 : SUCCESS in 44m 12s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/c019ec2a4c4744dc96cdcacec14e9ced : SUCCESS in 43m 58s","accounts_in_message":[],"_revision_number":3},{"id":"b07441fa0979b8dc5edcdc6792a00afe55300f09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 14:30:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"291f88a1e84b1c3f4c8d18c0a6a06fea276c5040","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 18:35:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cadc7176a1b4d25bb42234794187457\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63224d261ebc4adc8f89156a2872c72c : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/887b6901cbf34ba58644ec5c32a1b894 : SUCCESS in 11m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9fcb1c3915084293a895aef28cb7a6ed : SUCCESS in 16m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc86bd8589424ec5948be483d804f6ef : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1f24459181e4c88810e2d6d4b2eb99d : SUCCESS in 13m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/0cf19360d5144070886e4605f644704a : SUCCESS in 1h 03m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/28c0659af5634c828ebd5d3f60d8ef00 : SUCCESS in 1h 40m 51s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/250e7734c2764676b8a9a7665463a051 : SUCCESS in 14m 13s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/ff229ba8200f46b38a668212c4aaec1f : SUCCESS in 29m 18s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/02ac0c70b78f42d5a8181bac79666c8b : SUCCESS in 30m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e709ed4b33445a2b6d738d81c3945fa : SUCCESS in 49m 37s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/6747515f48034de4a91fa9fa0e3ae1b1 : SUCCESS in 40m 25s","accounts_in_message":[],"_revision_number":3},{"id":"c6fc1ae4f96136683ba8956a23000d17a3711ebb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 18:35:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d23b0097f720d94c959ef1def0a5e84e1c412345","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 18:37:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3cbb122fae24368accf6d45856f0cf4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b03819d5a904faa8276c0e538011016 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"845e5b24949bc0cdd021c9e9a0cf88bc74e7b3e8","revisions":{"4cf57574179fce140c26ab520edad3ad5045bfaa":{"kind":"REWORK","_number":1,"created":"2023-07-03 11:12:31.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/09/887509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/09/887509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/887509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/887509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/887509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/09/887509/1"}}},"commit":{"parents":[{"commit":"e05e3929d162182aaaa99c11fcf36f179ec6ffe2","subject":"sql: Fix incorrect columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e05e3929d162182aaaa99c11fcf36f179ec6ffe2"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-03 11:06:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-03 11:11:41.000000000","tz":60},"subject":"sql: Remove service_provider.relay_state_prefix default","message":"sql: Remove service_provider.relay_state_prefix default\n\nWe shouldn\u0027t specify a server default for a configurable option since it\nmeans our initial database schema is not consistently reproducible.\nInstead, we should specify the default at runtime. It turns out we\nalready do this and the server default was overkill. We can remove it.\n\nChange-Id: I74e47a9ed986c7c3af19676ac65f4d290bcb4cc0\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4cf57574179fce140c26ab520edad3ad5045bfaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4cf57574179fce140c26ab520edad3ad5045bfaa"}]},"branch":"refs/heads/master"},"b14aefd6a9a72729547ef201bcdeac7e143e146f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-07-03 11:33:09.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/09/887509/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/09/887509/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/887509/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/887509/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/887509/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/09/887509/2"}}},"commit":{"parents":[{"commit":"2bf70a10a2a7e0ef31f790c21d9cb6b2643c186d","subject":"sql: Fix incorrect columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2bf70a10a2a7e0ef31f790c21d9cb6b2643c186d"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-03 11:06:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-03 11:33:01.000000000","tz":60},"subject":"sql: Remove service_provider.relay_state_prefix default","message":"sql: Remove service_provider.relay_state_prefix default\n\nWe shouldn\u0027t specify a server default for a configurable option since it\nmeans our initial database schema is not consistently reproducible.\nInstead, we should specify the default at runtime. It turns out we\nalready do this and the server default was overkill. We can remove it.\n\nChange-Id: I74e47a9ed986c7c3af19676ac65f4d290bcb4cc0\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b14aefd6a9a72729547ef201bcdeac7e143e146f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b14aefd6a9a72729547ef201bcdeac7e143e146f"}]},"branch":"refs/heads/master"},"845e5b24949bc0cdd021c9e9a0cf88bc74e7b3e8":{"kind":"REWORK","_number":3,"created":"2023-07-06 09:23:32.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/09/887509/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/09/887509/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/887509/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/887509/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/887509/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/09/887509/3"}}},"commit":{"parents":[{"commit":"2bf70a10a2a7e0ef31f790c21d9cb6b2643c186d","subject":"sql: Fix incorrect columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2bf70a10a2a7e0ef31f790c21d9cb6b2643c186d"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-03 11:06:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-06 09:19:05.000000000","tz":60},"subject":"sql: Remove service_provider.relay_state_prefix default","message":"sql: Remove service_provider.relay_state_prefix default\n\nWe shouldn\u0027t specify a server default for a configurable option since it\nmeans our initial database schema is not consistently reproducible.\nInstead, we should specify the default at runtime. It turns out we\nalready do this and the server default was overkill. We can remove it.\n\nChange-Id: I74e47a9ed986c7c3af19676ac65f4d290bcb4cc0\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/845e5b24949bc0cdd021c9e9a0cf88bc74e7b3e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/845e5b24949bc0cdd021c9e9a0cf88bc74e7b3e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
