)]}'
{"id":"openstack%2Fkeystone~890282","triplet_id":"openstack%2Fkeystone~master~I05339b12b0475c80c465e1586e9990b1861ea8a0","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"35058":{"account":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"last_update":"2023-08-04 16:31:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I05339b12b0475c80c465e1586e9990b1861ea8a0","subject":"Use py3 as the default runtime for tox","status":"MERGED","created":"2023-08-02 02:51:30.000000000","updated":"2023-08-04 16:33:18.000000000","submitted":"2023-08-04 16:31:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890282","meta_rev_id":"3d6f628749bfbdb8979444b3e0550c54f91c6bc6","_number":890282,"virtual_id_number":890282,"owner":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-04 16:31:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2023-08-04 14:16:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2023-08-03 08:25:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-08-04 14:16:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-08-04 14:16:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-02 03:02:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-02 04:36:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-03 08:25:24.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-08-04 14:16:23.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2023-08-04 14:16:28.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"8054185b8a3895893c4e466369110dccf74c4832","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2023-08-02 02:51:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46449a8d3a4f36ba0e2c7797330bb50de5f9a97e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 03:02:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8470dd052084de789a79c842d7ca25c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/43118894f0cc4296934a61c97c0efd8b : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/88cd8e6539d94d64a3e39145557f6a0b : SUCCESS in 9m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb3f91cec63f0716026c52be250de5dab32a101e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 04:36:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4eec80f98ff4056be020dd74feff6ae\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2584e49316b249aca6d77a970b73a33b : SUCCESS in 5m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54e3a48966854e28a6094f1d9e255ec6 : SUCCESS in 15m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/478adf5485ca47d2baae0cb762ee8bda : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/147e294089934833b848d7af45dfee1a : SUCCESS in 13m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4cdc1d4943c14768bab1c0b86f346594 : SUCCESS in 22m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85ee4f67e7b64e2f8e1063df2990c624 : SUCCESS in 14m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f9c949c7d474bc3be485968640ee71d : SUCCESS in 12m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/48b9fe0cf0d24f1a9bd2c8640910254a : SUCCESS in 59m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/075bfeffd0f24fa48e6fb78a048e7409 : SUCCESS in 1h 33m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2a40e7c4fd049939d65505ff3ec6a6f : SUCCESS in 8m 47s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b2227c664e364bec925a0d9d3e1179cc : SUCCESS in 14m 49s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/bec080fad6484ba0a2d49e6b2e8d1539 : SUCCESS in 26m 28s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/496d0f8e4a874f89970eccd7c26c7fee : FAILURE in 43m 54s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/533c2169ba37417ab9949062a81188c9 : SUCCESS in 37m 22s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/e78b1d2137744c50b4c7f442b06708f4 : SUCCESS in 1h 00m 33s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/9fd3575b82144aa29f9e50a169fad0c0 : SUCCESS in 10m 49s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/6830491f96ea4d079c9ac72e6e0ea95c : FAILURE in 22m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/413aaf10e2b94a2087a7d1a31f42721d : SUCCESS in 1h 32m 49s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/f5fd57b853ca4785a691d921122169dd : SUCCESS in 44m 07s","accounts_in_message":[],"_revision_number":1},{"id":"312541dd1a409f11537fc6347259cc60cbb9f521","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-03 08:25:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ba4d8f431d5da1db2b2ae27c54ddefc38d91fb7c","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-08-04 14:16:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"53f739fe13d8f6398f1bcb1fc8713c9e9b999d0c","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-08-04 14:16:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"14e24e94338888d3ba4f7831fdc04e75443b7a42","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-08-04 14:16:39.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2416205f86bb80afd6111bf40839ebe5b664f919","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 14:17:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"849da7f026d4bc81bd63f5207f905ab79b8cf774","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 16:31:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c78e4ac36a6426f8f9a304d60d182a1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4819f9611bcb4d96b413af5c13382cfe : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83e9cb57533546809e65e99ae67075ab : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/620e269d24884f9bb46e865bfcba730d : SUCCESS in 15m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3d4f893cdbd4adfb80dbd4031f991c7 : SUCCESS in 13m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f91e266652594294a4fe9d0ab1fedf06 : SUCCESS in 12m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b15a359948b4c179f502928a6a00da5 : SUCCESS in 11m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/094b573ed6fa43739f6f54734fa7614f : SUCCESS in 1h 01m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ca78f283ee714ef7af1c2628103c068b : SUCCESS in 1h 18m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0326ad42efa1455593eb239409c924bb : SUCCESS in 10m 58s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9eef0b9f097a4ba6bd4048e04dc60769 : SUCCESS in 13m 13s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f7827cb0a5ef454295b2a019df183a43 : SUCCESS in 37m 39s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/4ff7d260b30e41bf9a8712238b59faa1 : SUCCESS in 35m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3f029d3ae944b47ae4fe04f04e0c21b : SUCCESS in 44m 34s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/f6c32268475b43cfbb0ed63ea0958e3e : SUCCESS in 41m 40s","accounts_in_message":[],"_revision_number":1},{"id":"b0f17368cd661010fe9df243dc2f87aff7776adb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 16:31:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3d6f628749bfbdb8979444b3e0550c54f91c6bc6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 16:33:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab344996e76f42e4adda512f150da970\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/b9c66ff317ee436e952876493867da16 : SUCCESS in 47s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4b9b48c1b84406c8935126b01019566 : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d3f8797e83744c3bec3549b2fb68735 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"38d9a1ebcaea22b607293af777bc5293c164b45d","revisions":{"38d9a1ebcaea22b607293af777bc5293c164b45d":{"kind":"REWORK","_number":1,"created":"2023-08-02 02:51:30.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/82/890282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/82/890282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/82/890282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/82/890282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/82/890282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/82/890282/1"}}},"commit":{"parents":[{"commit":"04e1b79cdde393382ffc6c7ab9462666a3ebe1b6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/04e1b79cdde393382ffc6c7ab9462666a3ebe1b6"}]}],"author":{"name":"niuke","email":"niuke@inspur.com","date":"2023-08-02 02:50:34.000000000","tz":480},"committer":{"name":"niuke","email":"niuke@inspur.com","date":"2023-08-02 02:50:49.000000000","tz":480},"subject":"Use py3 as the default runtime for tox","message":"Use py3 as the default runtime for tox\n\nMoving on py3 as the default runtime for tox to avoid to update\nthis at each new cycle.\n\nChange-Id: I05339b12b0475c80c465e1586e9990b1861ea8a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/38d9a1ebcaea22b607293af777bc5293c164b45d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/38d9a1ebcaea22b607293af777bc5293c164b45d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
