)]}'
{"id":"openstack%2Fkeystone~891115","triplet_id":"openstack%2Fkeystone~stable%2F2023.1~Iea95a3c2df041a0046647b3d3dadead1a6d054d1","project":"openstack/keystone","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-08-15 15:11:08.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2023-08-16 15:09:49.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-08-16 17:44:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iea95a3c2df041a0046647b3d3dadead1a6d054d1","subject":"Properly trimm bcrypt hashed passwords","status":"MERGED","created":"2023-08-12 04:49:42.000000000","updated":"2023-08-16 17:46:50.000000000","submitted":"2023-08-16 17:44:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891115","cherry_pick_of_change":890936,"cherry_pick_of_patch_set":4,"meta_rev_id":"3e45208140f5d384258ac7e65b852b4beb1bc1e4","_number":891115,"virtual_id_number":891115,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-16 17:44:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2023-08-15 15:11:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2023-08-16 15:09:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2023-08-14 08:43:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-08-12 15:35:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2023-08-16 15:09:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-12 05:08:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-12 06:22:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-12 15:35:16.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2023-08-14 08:43:54.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-08-15 14:07:21.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2023-08-15 16:10:57.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"6f30081466da48abc7eb82710c165f26738fd4e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-12 04:49:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"37cabbb6fa12cc1fcdbd4ff9b6d8d222255d6fc2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-12 05:08:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edd2936ccb524564b4eb1464db09a5f2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0eaa3923b91346c28aeb51352af12046 : SUCCESS in 12m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dc0e8ab5a36a4e84bf02fec381d29e28 : SUCCESS in 18m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"496780cca815b68467e25f8ac3c2608a47c31da3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-12 06:22:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47fbb0f547c64806aafd1c28c4b2539b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88c641f9dfba42b380db673893767a81 : SUCCESS in 18m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d71e7ae3941041e8a3823fbef629dafc : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ed5a578a983499db778c2d8d8f338f7 : SUCCESS in 10m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d7ef88ec6ba4fa89da943353c3fe434 : SUCCESS in 12m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8fac480d5a9481086bd33738ef62ee8 : SUCCESS in 10m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/2a40f93572f94ed8b428241f5ca886ee : SUCCESS in 1h 03m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c594cac428864f8f82c37c62f1cd7cc1 : SUCCESS in 1h 25m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/256b2d04e12e4685b7c53cff78258263 : SUCCESS in 6m 03s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/cd9bd7e84faf43b7afd27f6045fe129e : SUCCESS in 20m 31s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/ae0b83063dce45b3832a5a7a9ce54f7a : FAILURE in 41m 54s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2fb55c46ce74ffaab73c070326af969 : SUCCESS in 36m 23s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/a9ac9bb3f83247b188a77450a27a5adf : SUCCESS in 20m 18s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/e4d48e6996464738b40d2e5a2010847a : SUCCESS in 10m 28s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/305132c888b44932b2da3b6d40b29f9a : FAILURE in 15m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/957c791878c84e88901351808c9e6990 : SUCCESS in 57m 38s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/e1ff355d07074d9992188796dc4c1b83 : SUCCESS in 46m 07s","accounts_in_message":[],"_revision_number":1},{"id":"439f1c71c316dc20bdc645cdf7e96128b067de9e","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-08-12 15:35:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"250d8f1bdfb7c957a07a7f983b35b1e73fa120e6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-14 08:43:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b78d32a80caca09917fd084465f646f4b474f6ac","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-14 08:44:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"197a7cae4900bc08e559c82b9ca7c137da8553e1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-08-15 15:11:08.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9549f15596a3fa4a9edfc5d72d8f55b16c584e3b","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-08-16 15:09:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"51ff77993c7a82ba8d31a3446a20889e702eca3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 15:10:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f01b8d12cf3838f2a484dbc1c66b7d7f00646c45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 17:44:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2506280de08e4e13ac5270cdc93d3d0d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec3bc4ce523c4f71a96e9d820509030d : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/840cde2eeb1e41f5bfd0553fabe2d3bd : SUCCESS in 12m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6510b8ca11c4fcf8cfb15095836fe97 : SUCCESS in 11m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9b87c9b661447689081d836ebdd3375 : SUCCESS in 14m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/c984a01c197543c8b39b41d108e012dc : SUCCESS in 46m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0884483e256b4da2ab5620e745ec3369 : SUCCESS in 1h 14m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/766312b745074190bdf536e16c2ade76 : SUCCESS in 9m 19s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/5c6126af785e461a990d83594237dc8e : SUCCESS in 36m 51s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/ad148f7621e24119bed1aa3f4e34be7a : SUCCESS in 37m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef4c1f7963354593acad370a644c9a28 : SUCCESS in 44m 28s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/e2c4b215f30a49f88913cdf1e8a9fdcc : SUCCESS in 35m 46s","accounts_in_message":[],"_revision_number":1},{"id":"fcfbdae314a4e0733d7dc60a8cd13d4c4819a457","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 17:44:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3e45208140f5d384258ac7e65b852b4beb1bc1e4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 17:46:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc96829613aa4725b2b8f9574e8f993e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a08cc4483ecc4a9a95d8277ec41a2f2c : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7195ac70e02446d08d7520f88db14601 : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"df54af90d03b14ebcd6e662bc8ece1fc52ea7c1d","revisions":{"df54af90d03b14ebcd6e662bc8ece1fc52ea7c1d":{"kind":"REWORK","_number":1,"created":"2023-08-12 04:49:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/15/891115/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/15/891115/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/15/891115/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/15/891115/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/15/891115/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/15/891115/1"}}},"commit":{"parents":[{"commit":"a72ec5d6020c73751740f5ac75619a68cbab60f8","subject":"Merge \"Remove Dependency on Cryptography \u003e\u003d36.0.0\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a72ec5d6020c73751740f5ac75619a68cbab60f8"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-08-09 18:41:05.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-08-12 04:49:42.000000000","tz":0},"subject":"Properly trimm bcrypt hashed passwords","message":"Properly trimm bcrypt hashed passwords\n\nbcrypt  hashing algorythm has a limitation on length of passwords it\ncan hash on 72 bytes. In [1] a password trimm to 54 symbols has been\nimplemented, which resulted in password being invalidated after the\nkeystone upgrade, since passwords are trimmed differently by bcrypt\nitself, as well as len(str()) is not always equal to\nlen(str().encode()) as trimming should be done based on bytes and not\nstring itself.\n\nWith the change we return a byte object from\n`verify_length_and_trunc_password`, so it does not need to\nbe encoded afterwards, since we need to strip based on bytes\nrather then on length of the string.\n\n[1] https://review.opendev.org/c/openstack/keystone/+/828595\n\nCloses-Bug: #2028809\nRelated-Bug: #1901891\nChange-Id: Iea95a3c2df041a0046647b3d3dadead1a6d054d1\n(cherry picked from commit 6730c761d18aa547998f2add833c13f45f257fe7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/df54af90d03b14ebcd6e662bc8ece1fc52ea7c1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/df54af90d03b14ebcd6e662bc8ece1fc52ea7c1d"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
