)]}'
{"id":"openstack%2Fkeystone~891116","triplet_id":"openstack%2Fkeystone~stable%2Fzed~Iea95a3c2df041a0046647b3d3dadead1a6d054d1","project":"openstack/keystone","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-08-17 09:00:24.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-08-21 14:08:23.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Iea95a3c2df041a0046647b3d3dadead1a6d054d1","subject":"Properly trimm bcrypt hashed passwords","status":"MERGED","created":"2023-08-12 04:49:59.000000000","updated":"2023-08-21 17:28:52.000000000","submitted":"2023-08-21 17:27:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"891116","cherry_pick_of_change":890936,"cherry_pick_of_patch_set":4,"meta_rev_id":"cc86178fd285debb529f3071de7d557055e1a6df","_number":891116,"virtual_id_number":891116,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-21 17:27:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2023-08-17 09:00:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2023-08-16 15:10:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2023-08-17 07:52:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-08-17 09:44:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":1,"date":"2023-08-12 15:35:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2023-08-17 09:00:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-12 05:37:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-12 06:20:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-12 15:35:24.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2023-08-14 08:44:09.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2023-08-15 14:07:32.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2023-08-16 15:10:20.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2023-08-17 07:52:37.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-08-17 09:44:26.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"}],"messages":[{"id":"44d6479774a49f1be2c8173e3dab29b90537adb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-12 04:49:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b24b1119b45b8a44901590866d3a8407d902a5d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-12 05:37:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69a5c201ca7e4b609a1671addc62ddf7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/16eac45d5bcf44588fd66ab76b75978f : SUCCESS in 17m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a27c1f0ae7ea49d3b0c05939c74f16a9 : SUCCESS in 44m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c6333e4126d44c8b294f994c922c90f4a35cc4f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-12 06:20:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d980b97cc814501a4489cc634c64f6e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/859788188a6f4b5ea6cec26ca366c601 : SUCCESS in 12m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae0796d10ea94611a3e44228b239d0d4 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72ae0b3b37a14f06815da8ab9b6b41cd : SUCCESS in 12m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/567829a43d5b4c49b9cf1a9230989953 : SUCCESS in 13m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b3515ebd46d4fdbb6eb788ae3d4ce48 : SUCCESS in 11m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff563d7d4cb64630845ada8d2c5b71ea : SUCCESS in 13m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/f8f604c8442a4b11a5f20c8320661550 : SUCCESS in 1h 02m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0132caee59d9486d8b105160bfdf57bd : SUCCESS in 1h 23m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99f1b75a3f954dd291f9235cdc4e6693 : SUCCESS in 7m 18s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/146a325f44f74add98a93a7362a7d3bc : SUCCESS in 42m 02s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/71c39da38c8c48d7833d3c92bf87fe7b : FAILURE in 24m 54s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c07c043752504ec8aa5a77d9b80643f8 : SUCCESS in 33m 51s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/f54d5b215c4041b6b07464b1281e06f3 : SUCCESS in 38m 00s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/ff36742f39524376be6d57091bc17e9d : SUCCESS in 10m 49s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/99dd022d1a5f4b6bbfc5f55f90f15491 : SUCCESS in 26m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c1723f6327840528de45f15d8540ec4 : SUCCESS in 1h 01m 26s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/514fb014658d41c68c7d28d2cb6731d0 : SUCCESS in 27m 30s","accounts_in_message":[],"_revision_number":1},{"id":"183388a6d01a18acd36e533df200f7d30c519fff","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-08-12 15:35:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bf275c4cf59a818077c9a87266d0715a374b6ccc","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-14 08:44:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2723d163bca618ee7ab47a745023c24701d0d310","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-08-15 15:14:10.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1c5aecbcbcf43dc02574d797dac7d4737c3fd1f6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-15 16:13:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1a978d2cc7eeba14d2f1584b18fbf916a81246c4","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-08-16 15:10:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ad368cd672af91867c1ce12025bd233a6cdb4507","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-08-16 15:29:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b8e0ee236121946d716fe020f7ec590ffb64a9b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-16 15:58:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d549ffcddbb3a8527b68747bbb6ced5fcfc65f42","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-17 07:52:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8b7a19bd5ff77f1337ee9ce61c390d27e4ed915f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-08-17 09:00:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0e5ab2204707829c876c03fbb6ab5f68ba676fe4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 09:01:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"20914b3e27266ec0e6491b52187f81147a46fd6e","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-08-17 09:44:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5f0808f499d4b4bf2bdc9d3465a33533854ebe0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 10:32:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8594c497f582430fb2a67c86f5757fb2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0187c305574478390fbf3892efc1793 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4870d93ad18744e0bc36e90206b94a7e : SUCCESS in 12m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f240c79cfe0a41869a903a4bd1f14d8e : SUCCESS in 16m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68caacde7f4c49738d7dfc1de7cf56bb : SUCCESS in 12m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/5a88f93562fd402db1d3cccf41fedb65 : SUCCESS in 58m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/61149ebe6b0043c89f16250b9aeee65c : FAILURE in 59m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4557f861d3204433a040cea86b797315 : SUCCESS in 8m 33s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d043cd2d89a944cfa93fab19f5036607 : SUCCESS in 25m 38s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/a1ac2413b41844aea4688e19b105627f : SUCCESS in 27m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07b7796ffae14d4695690f9b25c92d94 : FAILURE in 51m 10s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/d3ee5d8f68374c3a849deebf3d4ac74b : SUCCESS in 43m 21s","accounts_in_message":[],"_revision_number":1},{"id":"1643297192842d0f339f8f0c1c2fb39651bba822","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-21 14:08:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c738a74b0a1aa64e71771f7184fe94d6ec3422d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 14:20:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1dfa8836ab2437492fab5a4979c5326\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/342a35b75f6b486fb5cf5d3acb0441a5 : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/28ba961a7eeb40349bc94f4333df9a0f : SUCCESS in 10m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ea3a83b8ee344d67ed3da15c0f7f972bddf43888","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 15:42:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56a67b3ac1ba4401ad53d0e1451ea4ff\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c5018b8585a468eaf9f613cbb71d8df : SUCCESS in 14m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/075ffe3e1856457ca3260d08057422f4 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4dfcb5a615114009a45195b465d06d33 : SUCCESS in 11m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2810a2f12e948d3a58ef679cb5c0f28 : SUCCESS in 12m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae900aa929e448f399325bf07a5894f8 : SUCCESS in 11m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a81da571dcd9492e81ae2af5df351588 : SUCCESS in 11m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/eee63b9a9a6f4270958fa50da87b8a74 : SUCCESS in 50m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d3211e7546a43b88c13fd6cba95fcfa : SUCCESS in 1h 26m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa21c721033146e080edaa656233720b : SUCCESS in 8m 02s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/b9199a3bccee4dde97093bf11e6539b2 : SUCCESS in 24m 19s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/78b7fcd4297c4878bd469c40b8bb00f5 : SUCCESS in 30m 54s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d47251395cab4d89be82bb7850c8fb09 : SUCCESS in 33m 30s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/03fff845afb04ccbbb74a8462ae06fb5 : SUCCESS in 38m 39s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/836880c674f644b5a70e2465a4dc85ea : SUCCESS in 21m 10s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/68f270be747a40e9a5d76113d614b433 : SUCCESS in 45m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab8bf7136e2d4ae48424a20c632bfeb1 : SUCCESS in 52m 12s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/91a32a6dfe5f40059830c4cf958df827 : SUCCESS in 33m 06s","accounts_in_message":[],"_revision_number":1},{"id":"5a4c5a031a04dbf7d2569cf6368eec8b73caa5bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 15:43:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"04a7bcfd79ab5f786d1a04397c396e4cee71312f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 17:27:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7eca69d2b45d4a85bb104dca4fbee6ef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ee8c007bd624f799f7bc8c92650e63e : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47dbd78078a14033a52ba7728c6ae301 : SUCCESS in 11m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63d9bfc50ec24a03bc4b01411da28601 : SUCCESS in 13m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3e332b31e224926927e06916c7f0f13 : SUCCESS in 14m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/2e98514991c6436f8d39419f191d16b5 : SUCCESS in 1h 13m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2461771ffc484d51a1eb26c7f965df45 : SUCCESS in 1h 22m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4df579959644e5388945e290b9570ac : SUCCESS in 9m 56s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d5c148f4f5ae4e6b997e8791001c28b3 : SUCCESS in 39m 57s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/8b81e4ce99c148a481bcea289805cbaa : SUCCESS in 36m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88a370c853e2439baef52bf18b6d1694 : SUCCESS in 58m 46s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/51e8270cb932400bb9c4f6d211b8711b : SUCCESS in 35m 09s","accounts_in_message":[],"_revision_number":1},{"id":"388c8820de4876884a2faa9acb0bf0db4555185c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 17:27:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cc86178fd285debb529f3071de7d557055e1a6df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 17:28:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f29e685cd2a4627a6ca94b155880c3c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b94bc45430b4bf7b288c69700969075 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da70883bc2314d32a8c9fa7186d97744 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"65f1fb6b4a54386f473369b05c8d10d77fb6710c","revisions":{"65f1fb6b4a54386f473369b05c8d10d77fb6710c":{"kind":"REWORK","_number":1,"created":"2023-08-12 04:49:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/16/891116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/16/891116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/16/891116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/16/891116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/16/891116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/16/891116/1"}}},"commit":{"parents":[{"commit":"1b3536a7a4d72e7f7b95cc1874a450accad3ec8d","subject":"Force algo specific maximum length","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1b3536a7a4d72e7f7b95cc1874a450accad3ec8d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-08-09 18:41:05.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-08-12 04:49:59.000000000","tz":0},"subject":"Properly trimm bcrypt hashed passwords","message":"Properly trimm bcrypt hashed passwords\n\nbcrypt  hashing algorythm has a limitation on length of passwords it\ncan hash on 72 bytes. In [1] a password trimm to 54 symbols has been\nimplemented, which resulted in password being invalidated after the\nkeystone upgrade, since passwords are trimmed differently by bcrypt\nitself, as well as len(str()) is not always equal to\nlen(str().encode()) as trimming should be done based on bytes and not\nstring itself.\n\nWith the change we return a byte object from\n`verify_length_and_trunc_password`, so it does not need to\nbe encoded afterwards, since we need to strip based on bytes\nrather then on length of the string.\n\n[1] https://review.opendev.org/c/openstack/keystone/+/828595\n\nCloses-Bug: #2028809\nRelated-Bug: #1901891\nChange-Id: Iea95a3c2df041a0046647b3d3dadead1a6d054d1\n(cherry picked from commit 6730c761d18aa547998f2add833c13f45f257fe7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/65f1fb6b4a54386f473369b05c8d10d77fb6710c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/65f1fb6b4a54386f473369b05c8d10d77fb6710c"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
