)]}'
{"id":"openstack%2Fkeystone~892757","triplet_id":"openstack%2Fkeystone~stable%2Fzed~I23f31bd85443784013a6aa158d80c7aeeb343993","project":"openstack/keystone","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2023-10-04 11:18:18.000000000","reason":"Change was submitted"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2023-09-08 14:26:11.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2023-10-04 08:24:18.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}},"hashtags":[],"change_id":"I23f31bd85443784013a6aa158d80c7aeeb343993","subject":"Add an option to randomize LDAP urls list","status":"MERGED","created":"2023-08-28 07:26:05.000000000","updated":"2023-10-04 11:20:18.000000000","submitted":"2023-10-04 11:18:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892757","cherry_pick_of_change":821086,"cherry_pick_of_patch_set":7,"meta_rev_id":"cc365a8319ed07124191ead56da66eff109691a9","_number":892757,"virtual_id_number":892757,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-04 11:18:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2023-09-08 14:26:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2023-10-02 10:06:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-28 08:05:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-28 08:41:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-28 11:31:33.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2023-08-28 11:31:33.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"f832da94f5f396ec2ae9a8d0b15cf17436243ecf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-08-28 07:26:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e8b9d92d0a711e7558319835431a77a05f89d71b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 08:05:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33f45b44bf0c49a2a4205e9bc67b1148\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e95237a1a1124d7bb2e76d34d4ffd8b4 : SUCCESS in 11m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/91d25daeed4c45a5bd16dc467cfc5365 : SUCCESS in 38m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cda4e212f7757f6565cb6007fb933de4d75fb424","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 08:41:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52eb5aa2836447dc81f483b0b9a06ade\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8c394aa01024142a36fa88d847af4d8 : SUCCESS in 12m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98e688f0d53449f692b83302fb1d749c : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27a66661bf8d472ab2c4ac079de61d4c : SUCCESS in 13m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c909b5c7bcc42f4a76fd4f7371cf739 : SUCCESS in 11m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f243885a963d4e3b82290c11a040d4eb : SUCCESS in 11m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48e649de340e4c708e22872aa9e48e66 : SUCCESS in 11m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/d6449b40bffc47bbb48ef5a87de99233 : SUCCESS in 58m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9a5a06e4085e4588bdc81983b008c56c : SUCCESS in 1h 14m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0ee5bb0f10b4ab3a893ae44609b8bab : SUCCESS in 8m 38s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/8151589e88394bf89a47c0f0d0fcdfac : SUCCESS in 28m 34s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/c2d73f4a4eea4d5094160fd6a1f568be : SUCCESS in 33m 01s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6b4f06f9a4047959624d3a12c176795 : FAILURE in 27m 29s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/0178afe4629344689d93a630b9bce977 : SUCCESS in 29m 59s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/b63a9ca16e4d4af3bb614e824830dad9 : SUCCESS in 12m 54s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ae802cf36bd840d9b8331fe08d42473d : SUCCESS in 31m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d4463564f1d451cb368c13dae9c1151 : SUCCESS in 1h 10m 30s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/04a680e757264bd3a647a634a34cbdf7 : SUCCESS in 34m 50s","accounts_in_message":[],"_revision_number":1},{"id":"00e95f10d042f2ddfde0a5e9045bef9e6ca94804","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-09-08 14:26:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ec593eaeb7c95fae53b04134b8679a37a64aa610","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-10-02 10:06:04.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"54c661af896ee57296054fc60630e40b4788e54e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 10:06:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"197cfd6ccdef01529cc6d6bad9ee729460790974","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 11:24:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/be30d6f16c134b8e8b9fc823517ddb69\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c7a3b73db7d465793cd4ef0eb384161 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/806ca15cf9ff43cfbe2bc296a85096e1 : SUCCESS in 11m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f1f7554ac2c4306a32941c382621078 : SUCCESS in 12m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b64b6eef1a734e30a646854810002698 : SUCCESS in 11m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/8a1fb3f414c74705956419a91a4dd574 : SUCCESS in 1h 02m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3459d25f776f41f1a2a3207bdbd5c5ba : SUCCESS in 1h 10m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55726c6c0ff146f595296d47531bdc02 : SUCCESS in 10m 42s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/7aa9350f9062444eae721aba88581f4e : SUCCESS in 37m 15s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/5ac8016425414ea1b27b251e878f52e7 : SUCCESS in 33m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f21d6fd3bc184d1ba297d20794ca342f : SUCCESS in 46m 08s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/03a3eb8e482d40eb91752db7797e9c5c : FAILURE in 50m 56s","accounts_in_message":[],"_revision_number":1},{"id":"687eb1967196ed8f213c6e7c47ce14f74dbda826","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-10-04 08:24:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"99809d421d502fbc4d1816bfb9035703f573410d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 09:38:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ebbb4f2f2364882af550ae25c26cb3a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e2a8da8f70845b0a465c411843fc2a2 : SUCCESS in 14m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0861e51f18364580ac2a53f7de0262d0 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d504e34150a455dab92d1e62de42c4c : SUCCESS in 11m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7fc060c46b74bf4b72f532dae1379e3 : SUCCESS in 9m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e67707b3dc2948d7b143f517c19095bd : SUCCESS in 10m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b729bbb03de4294810b5c8ddebf9c3f : SUCCESS in 14m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/2153b8b3a03d4265a02c64ee392e0483 : SUCCESS in 59m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c217c2aec4b141849c605ef8619ede44 : SUCCESS in 1h 06m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cce5849a39244529df20501155761bb : SUCCESS in 9m 22s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/97cb6e4ab3ee43478ebd9762693c1abe : SUCCESS in 32m 42s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/0f3050f62f404a10aa04de13bdbc5dd5 : SUCCESS in 46m 12s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/507a7972152f428bbde204bcef9afa7c : SUCCESS in 43m 05s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/91af711330ac4e7583296e1f34b49484 : SUCCESS in 38m 26s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d4a461244fab44a4a7e6704f7461406f : SUCCESS in 24m 49s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/2eb24db3d0d04c5580545c743aeecf84 : SUCCESS in 42m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b670e567e3e64d44888dc558b57d64dd : SUCCESS in 54m 42s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/d829a194101b47b6b0d8922c5aa00ea3 : SUCCESS in 42m 04s","accounts_in_message":[],"_revision_number":1},{"id":"19478ebab98d4abf4badbce5a2bc889001b2226d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 09:38:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"958b601a289645566b2ce44dd08ba2c0fb108a44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 09:48:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3e4732515b846eebae1faf8feae44a2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/46e38b8a62a94386b357ecc7b1692443 : SUCCESS in 41m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4454e1691b4849d28ebe9c1739b994b3 : SUCCESS in 50m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f439a9f75177225e7b1da1bb1dd8862521b29553","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 11:18:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c73b26309dee4e84ad7bf4652a115859\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b37931e9f9a434d9f81f3953329695b : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b622d13d58d4816938ecc6106524545 : SUCCESS in 11m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16698980b8294330839e6fb29a105623 : SUCCESS in 11m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7beb3dcbcc9441abb180c612ac5c28a : SUCCESS in 11m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/407c13d8389b420f87fa1bc9a499154c : SUCCESS in 1h 03m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2ae64b1f6e25468c84a433f922c7986d : SUCCESS in 1h 09m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9166db61f5ed4d44a50e34b0f7d938e1 : SUCCESS in 8m 01s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/316c40cf3a25430c8c429432cc4ca791 : SUCCESS in 35m 28s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/d10ada92f332432da9cc1dfccf5a5444 : SUCCESS in 45m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f8b3686b3db40e798e238a66bbe0c3c : SUCCESS in 39m 51s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/3a26959eefdf47da8f640136b8e12136 : SUCCESS in 43m 03s","accounts_in_message":[],"_revision_number":1},{"id":"297a04950c7715165f96b5136bd97d55d7c03e31","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 11:18:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cc365a8319ed07124191ead56da66eff109691a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 11:20:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dba329b032bc417593397e39b6da1db0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f2bc3918d5946d19d419523733a22a6 : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/541d5e88c95644adb82637c2ef9f5f0c : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"72a4fc0f3ccf7a5ca9fc40e5364e14f881ec27b2","revisions":{"72a4fc0f3ccf7a5ca9fc40e5364e14f881ec27b2":{"kind":"REWORK","_number":1,"created":"2023-08-28 07:26:05.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/57/892757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/57/892757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/57/892757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/57/892757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/57/892757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/57/892757/1"}}},"commit":{"parents":[{"commit":"0f6b645ce8ffbca9a33c5ab971a46941e96b8c1c","subject":"Merge \"fix(federation): allow using numerical group names\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0f6b645ce8ffbca9a33c5ab971a46941e96b8c1c"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-12-08 13:52:35.000000000","tz":60},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2023-08-28 07:26:05.000000000","tz":0},"subject":"Add an option to randomize LDAP urls list","message":"Add an option to randomize LDAP urls list\n\nSince LDAP is now readonly, the current behavior might be\nunexpected. By randomizing the list, we assure a more gradual\nfailure scenario if the first server on the list (as specified\nby the user) fails.\n\nChange-Id: I23f31bd85443784013a6aa158d80c7aeeb343993\nCloses-Bug: #1953622\nResolves: rhbz#2024602\n(cherry picked from commit 36d57d2a83a2a4a905337239d1a57d2e6a61c842)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/72a4fc0f3ccf7a5ca9fc40e5364e14f881ec27b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/72a4fc0f3ccf7a5ca9fc40e5364e14f881ec27b2"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
