)]}'
{"id":"openstack%2Fkeystone~901886","triplet_id":"openstack%2Fkeystone~master~I163f0327ede2a720c3b800dc4757d4791ed47d00","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2023-12-01 14:58:48.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-12-02 11:46:30.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2023-11-27 10:20:37.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}},"hashtags":[],"change_id":"I163f0327ede2a720c3b800dc4757d4791ed47d00","subject":"Drop compatibility code for Python 2.y","status":"MERGED","created":"2023-11-26 06:19:46.000000000","updated":"2023-12-02 14:40:13.000000000","submitted":"2023-12-02 14:38:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901886","meta_rev_id":"18214d4bf30995db63ee6dd69763c6d2c3439f80","_number":901886,"virtual_id_number":901886,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-02 14:38:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2023-12-01 14:58:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2023-11-29 03:43:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"value":1,"date":"2023-11-29 06:08:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-11-29 03:43:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2023-12-01 14:58:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-26 06:36:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-26 08:12:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-26 10:33:20.000000000","updated_by":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"reviewer":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"state":"REVIEWER"},{"updated":"2023-11-27 06:07:03.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2023-11-27 06:07:03.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2023-11-29 06:08:44.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"0fa6f5c81beb0084c5ede8b9ee8c7c4fa267ddf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-26 06:19:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33e73b4e975155611cf2527f3a51fcd2af5bba81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-26 06:36:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b0e0147753e458390ab131e6d13c80e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/171e40d9aa2d4aa1be2a40867f0528df : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b508df5c29040f2bb1207303547503a : SUCCESS in 15m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f9bad1aa6d66149be0efacb2247054b0d4c4ba8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-26 06:59:52.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5a9393001a3e3b5fd8e06ab0bb0a97829c143498","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-26 07:11:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cf4b626b160428cbe7a6d7c047fca9c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5ed6bba68f884cd7949328f0bcfac73e : SUCCESS in 10m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b0c7d4818faa44df8518d5b70fcaf8cd : SUCCESS in 9m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8f0eab36dc6760b1af193bf5fceda3bf2a8f36ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-26 08:12:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b38cee772c24b4e82ad96e3ccd58ba3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8684ab2edc448928325eaec7f888f3d : SUCCESS in 15m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5787653b9dc48c297e9b97c281cfe02 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7e64f2ed84b4178afb968bc30b8c2ee : SUCCESS in 11m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6175df76d913460c921f429478160dab : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/558cb14895e74677a9916679c79162d2 : SUCCESS in 12m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/8738789c9e13447c86e130cfc9c83f24 : SUCCESS in 38m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8961f83fd0f422396720b4d5a0d5660 : SUCCESS in 57m 12s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/134a4c45f60a446f8125a943fcd0a493 : SUCCESS in 12m 23s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/0964944b7a17489ca0f35f9b2c5ffd63 : SUCCESS in 18m 57s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/a1fc830d50244b9b90bd26e4d011c8bc : SUCCESS in 46m 01s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77d65cc8114447198cc54b40ae2b0e30 : SUCCESS in 19m 24s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/d9e2b3e8af65418688a6447944823a61 : SUCCESS in 34m 02s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/29da4648734f4ab594728aeb6ad23323 : SUCCESS in 21m 01s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/435275ed770f4578a3a20d628852e44c : FAILURE in 10m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30b2d70f74c3490da64419e65edaf8a1 : SUCCESS in 1h 06m 15s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/70a4c37fc61e457d9af16dfa48d07c89 : SUCCESS in 48m 48s","accounts_in_message":[],"_revision_number":2},{"id":"85b5feb20354c3481a456c9e82c5c1f0f160820f","author":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"date":"2023-11-26 10:33:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9d9ba7a2861b236b3b2ca2f8264a7103be2bd423","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-11-27 10:20:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"92e32b420af071cbd56f6c1a46a9f577dc504056","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-29 03:43:14.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9877f02dd2dd3b46ab85a09892b21c3965a059ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 04:46:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ba3fcb76ee4429895cc78f0b7951ee1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f377fbefa0d4eee93c7d798e3afb99b : SUCCESS in 11m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/60209e71cd62497a92f9aa42dfec815a : TIMED_OUT in 1h 01m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ccd3446af3505e1f841552857e9475f035d9addc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 05:05:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffb58dcbf0114b6a96b1ca3670736fc6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78f77644490549419c9ed9390b1ba401 : SUCCESS in 15m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b679ae97e1f94e55a34e43aea13855ea : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a08ccd01139542508ae27640ae033ede : SUCCESS in 10m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8bb4e75e900644e799ecef016326867d : SUCCESS in 11m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a14e4e2d81834c5a9a33067632c32f77 : SUCCESS in 10m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/e35f14f9ec0a495aa0a9db82a059a537 : SUCCESS in 54m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/53d89d9fc69540de90375dcc6826422b : SUCCESS in 1h 05m 29s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/710cd24ea5104fee89857b4eb5889ed0 : SUCCESS in 13m 38s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/59281aaf7d78481d9e4c3624a18c4bb0 : SUCCESS in 34m 12s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/42c5fa9a2fbb4a7bb619b83455761f3b : SUCCESS in 50m 24s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c39eb9bf3fe544d68ec711d5f87a1853 : SUCCESS in 44m 32s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/3fe8c757380f40adb50bd4c020788f1a : SUCCESS in 28m 32s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c6aac0aa83674900b2bde2c76ece026b : SUCCESS in 23m 32s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/eb2f42fa45b7468f8b3cd22c32e427f8 : FAILURE in 23m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8d80ba238e24fa0a66ee1e7f0b9aa3a : SUCCESS in 1h 12m 32s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/46a5f029bb5a4f41aab7397d6456e54d : SUCCESS in 42m 14s","accounts_in_message":[],"_revision_number":3},{"id":"bb13cafe83ad59d80f8b512c894e8bc39bf64281","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-11-29 06:08:44.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"523b735835b1979c624fdfd4489bbf5499434fec","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-12-01 14:58:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1a1cfdab2fd1719fe4addecfc8af1947097390e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 14:59:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"103ae65ac4d180c94399fa659ed13b7e8662ed2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 16:32:42.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e466c0f60f3d4988a5d8d17bf27a9d60\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/416c6413b88949678854111f9f03c370 : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5b1d95bf6524a6f9874f3929c2ff75b : SUCCESS in 11m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e0411c9b10c7408985db92e4732f7cde : SUCCESS in 12m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bcbe4a294784d93ba7fa6a4dfcc0e6e : SUCCESS in 11m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/48f9278f625f4ed3bfafae4e9784e507 : FAILURE in 56m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a38e563b5140492196b4bf496461cd18 : SUCCESS in 1h 17m 05s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ef73d5e1efbb4884b39955e27a24aa5e : SUCCESS in 13m 32s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/e67520d60db44f22bdf354ee3ae173cb : SUCCESS in 26m 31s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/91d1f5d6728f402bab06975433b57657 : SUCCESS in 26m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd98f1da148e4cd49a9e800f8d312a94 : SUCCESS in 47m 24s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/6ebbd9f6691c4285a65708838dcfec96 : SUCCESS in 37m 58s","accounts_in_message":[],"_revision_number":3},{"id":"9ec14af5566b059a2143aa744b31347dd7b90459","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-02 11:46:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fc58eae817433bbd791d22149df56fcf6dff4699","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 12:25:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2709524248ac42b795c41d9ffe825bf7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/476d0e7fb9c640d6bcf6ca846dfbe3d0 : SUCCESS in 14m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/33403a29577048ba89ade52b6d78610f : SUCCESS in 36m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"08ebb35098205fbde1ff8b6bf1c69456eaeabe1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 13:03:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6419a63383864c74b7945d4096976b54\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88e32f0928574ce298bd88badde850b4 : SUCCESS in 14m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa0ffcee8cf84dde9b75f63122ff4528 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37a0d4b246684be7ae57547dd2f23b97 : SUCCESS in 9m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6994f55d9f1947a491b6ae779db37f5c : SUCCESS in 11m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81effc9818a046ffb650127cc9eb2e58 : SUCCESS in 9m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/74a025e6e1ff4d849dc0c669b6f611d1 : SUCCESS in 1h 07m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f7741184bb854b6ab5f3288577fedcf7 : SUCCESS in 1h 01m 30s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0e9bd5044c464e9b9cb511395f796208 : SUCCESS in 14m 03s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/43eda094e37e465a88736c710224727c : SUCCESS in 19m 49s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/ab847dc5bab34198978e95853639cd62 : SUCCESS in 47m 07s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e85196d9599748f1923a0fe8aa6baf3d : SUCCESS in 36m 51s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/c7a5a9a9b27949199c7922d9b4a87307 : SUCCESS in 19m 45s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/81a8140c978a4e99b669edc516cc7347 : SUCCESS in 21m 16s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/eb8039f259044ba28c4f754ebbe9dea0 : FAILURE in 10m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/383bc1c9f337419ea450ad64b0c2f78f : SUCCESS in 1h 00m 21s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/da502cd935be4bae84bda3c78f0aac93 : SUCCESS in 27m 09s","accounts_in_message":[],"_revision_number":3},{"id":"f9052d9155c1152423a9336da1c58425aec3b6e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 13:04:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0d18a34a0a5011ab59222cf64934e96f5f305545","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 14:38:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43ea170400e743e8ab5dae965c155882\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7642571e73394f1abb2d5decad83ffc1 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87a2e177dba7476982bf9b9d96e329b7 : SUCCESS in 10m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/65e70ea117114f1a8f9951b920ef4655 : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d455cb8ae4a42f9ad1eba664e3134c1 : SUCCESS in 9m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/cd3d7c58d4ad485aaa8de8a1d55ecb69 : SUCCESS in 39m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c26a04b252c14a00bf77e9c84397f881 : SUCCESS in 1h 29m 35s\n- keystone-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/02915b2e284546da8b6e53c1977f044a : SUCCESS in 13m 19s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/4be8e28f69434eb9b643eb335309ea50 : SUCCESS in 37m 59s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/910eec2f0f06434394701a2dc53ba616 : SUCCESS in 34m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/844b326125484db8a1b8d78ddad31471 : SUCCESS in 1h 01m 48s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/dfa40b64fa9945c5951c3870810d98c7 : SUCCESS in 49m 32s","accounts_in_message":[],"_revision_number":3},{"id":"a455a2ba19eaf4725bd337614b65e3afb4b1c5f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 14:38:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"18214d4bf30995db63ee6dd69763c6d2c3439f80","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 14:40:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b8e37bb6c7949a99b6f1e602b4d9e9a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b9840667d43417f8943f4d847f1ed8d : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6458a8f81572cc13f8bc77527db73329b46af532","revisions":{"1f28e0a7d0f63fa76a16c7047e7c62267c9d2c92":{"kind":"REWORK","_number":1,"created":"2023-11-26 06:19:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/901886/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/86/901886/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/86/901886/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/86/901886/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/86/901886/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/86/901886/1"}}},"commit":{"parents":[{"commit":"1a21fbae2616a70cf3b912cce2ee6487e1b70161","subject":"Merge \"Add default manager role support to bootstrap command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1a21fbae2616a70cf3b912cce2ee6487e1b70161"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 06:18:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 06:19:35.000000000","tz":540},"subject":"Drop compatibility code for Python 2.y","message":"Drop compatibility code for Python 2.y\n\nThe inspect.getfullargspec method is available since Python 3.0.\n\nChange-Id: I163f0327ede2a720c3b800dc4757d4791ed47d00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1f28e0a7d0f63fa76a16c7047e7c62267c9d2c92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1f28e0a7d0f63fa76a16c7047e7c62267c9d2c92"}]},"branch":"refs/heads/master"},"cb6692bf1f12cfb97a31e2dbfc3ae0c15e56ae08":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-11-26 06:59:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/901886/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/86/901886/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/86/901886/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/86/901886/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/86/901886/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/86/901886/2"}}},"commit":{"parents":[{"commit":"5fa79e9aa4835b25d7367283cebe1bc1ad5475de","subject":"Fix bindep.txt for python 3.11 job(Debian Bookworm)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5fa79e9aa4835b25d7367283cebe1bc1ad5475de"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 06:18:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 06:59:52.000000000","tz":0},"subject":"Drop compatibility code for Python 2.y","message":"Drop compatibility code for Python 2.y\n\nThe inspect.getfullargspec method is available since Python 3.0.\n\nChange-Id: I163f0327ede2a720c3b800dc4757d4791ed47d00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/cb6692bf1f12cfb97a31e2dbfc3ae0c15e56ae08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/cb6692bf1f12cfb97a31e2dbfc3ae0c15e56ae08"}]},"branch":"refs/heads/master","description":"Rebase"},"6458a8f81572cc13f8bc77527db73329b46af532":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-11-29 03:43:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/901886/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/86/901886/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/86/901886/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/86/901886/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/86/901886/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/86/901886/3"}}},"commit":{"parents":[{"commit":"24b77bb6436f2f75c9555530915c2ec6be6f248a","subject":"Fix bindep.txt for python 3.11 job(Debian Bookworm)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/24b77bb6436f2f75c9555530915c2ec6be6f248a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-26 06:18:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 03:43:14.000000000","tz":0},"subject":"Drop compatibility code for Python 2.y","message":"Drop compatibility code for Python 2.y\n\nThe inspect.getfullargspec method is available since Python 3.0.\n\nChange-Id: I163f0327ede2a720c3b800dc4757d4791ed47d00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6458a8f81572cc13f8bc77527db73329b46af532"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6458a8f81572cc13f8bc77527db73329b46af532"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
