)]}'
{"id":"openstack%2Fkeystone~902032","triplet_id":"openstack%2Fkeystone~stable%2Fvictoria~I4fd7bf6b194c38815c2a9cdbab92a07315397eab","project":"openstack/keystone","branch":"stable/victoria","topic":"bug/1999068-stable/victoria","attention_set":{},"removed_from_attention_set":{"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2024-03-26 18:20:59.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4fd7bf6b194c38815c2a9cdbab92a07315397eab","subject":"Don\u0027t forget to check if authorization fails","status":"ABANDONED","created":"2023-11-28 08:43:48.000000000","updated":"2024-03-26 18:20:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":874346,"cherry_pick_of_patch_set":5,"meta_rev_id":"ba451befb0de9b03d8fb9a6d5bd93778ac7920e9","_number":902032,"virtual_id_number":902032,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-11-28 10:15:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-28 08:43:48.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2023-11-28 10:15:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-26 18:20:59.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"51692374266e182c496f1ea2e18280ad111c8f28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-11-28 08:43:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"35643a205b2277c3ccb9b7f8d3b728411bdfce18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 10:15:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9659ee7cd3374ee893df897bf5f25ad9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15b0afce5d5f495bb466681fce8a9bb6 : SUCCESS in 15m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef5b494d31f84b8a8985f57a1065cb09 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f18d5766ddd4d089c1140d1552ff568 : SUCCESS in 14m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f4803f4f63745fa8e43b074fce40e0b : SUCCESS in 13m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d897d1ff7b684bab84ad81f8e6d73501 : SUCCESS in 12m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/f4288b64782b4959bccd16c88a643067 : SUCCESS in 1h 23m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f17d7d952734347a2d23e885cf27ef2 : SUCCESS in 1h 23m 35s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/33ced82ebda14fc78bf2c4a4b88b3217 : SUCCESS in 31m 12s\n- keystone-dsvm-py3-functional-federation-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b5c4b0508ac4831b4cb5b250da756a1 : SUCCESS in 25m 27s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-focal-k2k https://zuul.opendev.org/t/openstack/build/aca46fa7c59045dd90f1800cf2b6ca1e : SUCCESS in 29m 08s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/4d83791d2e194263a0af0efd0dc450f8 : FAILURE in 11m 05s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/e843d427dec4492d87e2bbd639f015e4 : FAILURE in 13m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35c6b868b38c4b98b012f5e6774e938a : SUCCESS in 49m 35s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/2705e00ea85a472282067fac6f9d2675 : SUCCESS in 48m 28s","accounts_in_message":[],"_revision_number":1},{"id":"ba451befb0de9b03d8fb9a6d5bd93778ac7920e9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-26 18:20:59.000000000","message":"Abandoned\n\nstable/victoria branch of openstack/keystone is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/victoria if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f04a01b4cf086e517499590d68097e6a8a3fa246","revisions":{"f04a01b4cf086e517499590d68097e6a8a3fa246":{"kind":"REWORK","_number":1,"created":"2023-11-28 08:43:48.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/32/902032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/32/902032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/32/902032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/32/902032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/32/902032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/32/902032/1"}}},"commit":{"parents":[{"commit":"b59ab8e3362e1005c62f542f5614255a6f9c6c07","subject":"Merge \"Add an option to randomize LDAP urls list\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b59ab8e3362e1005c62f542f5614255a6f9c6c07"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2023-02-18 08:09:39.000000000","tz":-360},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2023-11-28 08:43:48.000000000","tz":0},"subject":"Don\u0027t forget to check if authorization fails","message":"Don\u0027t forget to check if authorization fails\n\nThe fix is to copy the missing check from class AuthProtocol\nof keystonemiddleware.\n\nCloses-bug: 1999068\nChange-Id: I4fd7bf6b194c38815c2a9cdbab92a07315397eab\n(cherry picked from commit 1544c7fc0aca9638c7cd31742e778d9327d2a62c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f04a01b4cf086e517499590d68097e6a8a3fa246"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f04a01b4cf086e517499590d68097e6a8a3fa246"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
