)]}'
{"id":"openstack%2Fkeystone~902034","triplet_id":"openstack%2Fkeystone~stable%2Ftrain~I4fd7bf6b194c38815c2a9cdbab92a07315397eab","project":"openstack/keystone","branch":"stable/train","topic":"bug/1999068-stable/train","attention_set":{},"removed_from_attention_set":{"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2023-12-21 14:05:11.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4fd7bf6b194c38815c2a9cdbab92a07315397eab","subject":"Don\u0027t forget to check if authorization fails","status":"ABANDONED","created":"2023-11-28 08:44:47.000000000","updated":"2023-12-22 09:32:28.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":874346,"cherry_pick_of_patch_set":5,"meta_rev_id":"03c0503cd5f46dbf077bed7433a69ede138b9892","_number":902034,"virtual_id_number":902034,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-11-28 10:22:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-28 08:44:47.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2023-11-28 10:22:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-22 09:32:28.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"}],"messages":[{"id":"a28f3865b8c0bb02874d2dafc2e50a7be15bc572","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-11-28 08:44:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"090030733a575f410c4a554ff011788a80619b1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 10:22:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/34e9cfe48827417a95ae217bdcc352b2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea9fa5d8436e418b895b32a01f0903ee : SUCCESS in 14m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0def4d613bbb46a4a79cfc015c97b407 : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bdf3a3670e1a461c9a4b050653807dc3 : SUCCESS in 14m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc04dc3c61e04865b9582aaf7b0af0b2 : SUCCESS in 14m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4eca2e6ed9454925a2e6c002c26b8876 : SUCCESS in 11m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73035fd4dc3a451aa5bb0c5d8aea4b4f : SUCCESS in 12m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/63fac90dd9f7487393b3c908e8aa9c9a : FAILURE in 16m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/525c7bb9479041779c71cf757a2c0b8d : SUCCESS in 1h 07m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/004c67501138432384c96363618f696d : FAILURE in 19m 42s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/007f8df60d584f539baab94fd557ea7a : FAILURE in 16m 45s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/b6d73a8e954a46e9b9ce906672827cb5 : FAILURE in 12m 39s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/921c75992a194527a537f816dce81953 : FAILURE in 16m 40s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/ea380c5c107d4a2c8348c19b229fab4f : FAILURE in 15m 14s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15-k2k https://zuul.opendev.org/t/openstack/build/cf4d8897a131491f90a99bd0a18b2885 : FAILURE in 6m 11s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/721cb6b3925b47678930bee7f8082dc1 : SUCCESS in 21m 35s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/7d59b3a6773a4774a7a7c2f973cef5d3 : FAILURE in 19m 02s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a2ebd244e68e43f78a90d2d87ecfd91f : SUCCESS in 1h 18m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/517c378c242140f7913e1d03341ece7b : FAILURE in 9m 45s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/7551ad0822154e2a92f463a10a5bb164 : SUCCESS in 48m 54s","accounts_in_message":[],"_revision_number":1},{"id":"130497810d8f2e017f284731d9583c924c14801e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-12-21 14:05:11.000000000","message":"Abandoned\n\nTrain is about to transition to End of Life. Open patches needs to be abandoned before branch deletion.","accounts_in_message":[],"_revision_number":1},{"id":"ac3e7b49285a73b6d50cf98030000b287b523a52","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-12-21 17:14:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03c0503cd5f46dbf077bed7433a69ede138b9892","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2023-12-22 09:32:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e035311214b1dc096b5773ccef515b1029732a05","revisions":{"e035311214b1dc096b5773ccef515b1029732a05":{"kind":"REWORK","_number":1,"created":"2023-11-28 08:44:47.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/34/902034/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/34/902034/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/34/902034/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/34/902034/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/34/902034/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/34/902034/1"}}},"commit":{"parents":[{"commit":"9d699a73fda748a6f280cbd61568b847316faa0a","subject":"Merge \"Limit token expiration to application credential expiration\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9d699a73fda748a6f280cbd61568b847316faa0a"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2023-02-18 08:09:39.000000000","tz":-360},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2023-11-28 08:44:47.000000000","tz":0},"subject":"Don\u0027t forget to check if authorization fails","message":"Don\u0027t forget to check if authorization fails\n\nThe fix is to copy the missing check from class AuthProtocol\nof keystonemiddleware.\n\nCloses-bug: 1999068\nChange-Id: I4fd7bf6b194c38815c2a9cdbab92a07315397eab\n(cherry picked from commit 1544c7fc0aca9638c7cd31742e778d9327d2a62c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e035311214b1dc096b5773ccef515b1029732a05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e035311214b1dc096b5773ccef515b1029732a05"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
